)]}'
{"id":"openstack%2Foslo.db~802585","triplet_id":"openstack%2Foslo.db~master~I92a7ccdd48eedd4c788384033743daf50a9dc113","project":"openstack/oslo.db","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2021-11-15 11:29:39.000000000","reason":"stephenfin replied on the change"}},"hashtags":[],"change_id":"I92a7ccdd48eedd4c788384033743daf50a9dc113","subject":"Don\u0027t call mapper() outside of declarative registry","status":"MERGED","created":"2021-07-27 17:30:19.000000000","updated":"2021-11-15 14:47:31.000000000","submitted":"2021-11-15 14:45:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802585-sqlalchemy-20","meta_rev_id":"50d46bf943390b0518056c2de1b1cd7aeaa416f8","_number":802585,"virtual_id_number":802585,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-15 14:45:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2021-08-24 13:06:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":2,"date":"2021-08-11 12:47:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-11-12 10:20:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-27 19:14:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-11 12:47:10.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-08-24 13:06:05.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"e8866b51f4807468f426d4309d48a915d8006d29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-27 17:30:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"26060181d993a014a3ce8df13d97213ed1066c1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 19:14:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4b7a77e4b3b74f099ab73ec2c0cf5321 : FAILURE in 1h 05m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21389a78071f46808759cab44ebd81d9 : FAILURE in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce6d73d671d3485781e7927537c0f52c : FAILURE in 6m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f163a63a305466f8670f8b7255ba7f2 : FAILURE in 6m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd69e702ed494412a980dafd171de358 : FAILURE in 7m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17eda88bcb454b34bd77eb6f2251cf39 : SUCCESS in 10m 58s\n\nWarning:\n  Comments left for invalid file oslo_db/sqlalchemy/engines.py","accounts_in_message":[],"_revision_number":1},{"id":"36dd8429a0948bc022ff5fdf0022539ffa686388","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-28 14:38:18.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"56fb38959aed2704990cd140a9e87228c06c93cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 16:00:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6264a5d8b0c14a308a3ee6263d91efc2 : SUCCESS in 1h 16m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/686524f0ad75402184cea33df878e2c4 : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5664ab1b25042cd81f5cd57fe16f664 : FAILURE in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df39b463c28146e381363d679ed0cbd4 : FAILURE in 7m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c21c1359f6a5496d97e2df42a09cd6f9 : FAILURE in 7m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7aa22cedb5647f4851cd3ad4caf3621 : SUCCESS in 7m 31s","accounts_in_message":[],"_revision_number":2},{"id":"cbed80ecbab5fe30d4cf99832bba345a3b00e34e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-29 10:44:29.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"0736c27f4e1162d5dc30b11067a16d579fa1fa35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 12:38:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/893d48ffa12d4ac3a848f462964bc69b : SUCCESS in 1h 43m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/013610eba52f45d98dfd39c57629216e : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f392be8860b47bea52011b50c61cf9b : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2fdf2242d8194971a2022e0804f3569d : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ba955bcbedb40e8a849889f99286a6d : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67fb2469fbaa4a1fb4d8782345bd3cae : SUCCESS in 6m 46s","accounts_in_message":[],"_revision_number":3},{"id":"edb95c802124477d40dfce5976e25a5e27cf78fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-29 15:37:53.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"9cd66362f12599fe6e81f7afe12a462790bd147c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 16:58:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a4aa3906e7f34aeeb957553f934453aa : SUCCESS in 1h 14m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52300696d82741f08234b773b6260d6b : SUCCESS in 8m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/160ead7056fb44219ca07c87f74e4692 : SUCCESS in 7m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be7f8176a7f945c3ae17a2e9a4001ea7 : SUCCESS in 7m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f93e722f98442908a35b0256df547df : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28f3bcbb8fcf4d2c8c6cbc3ad68da40d : SUCCESS in 8m 32s","accounts_in_message":[],"_revision_number":4},{"id":"69470cbbdd5e1204a5f2e2438493e3e414dfce66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-10 16:45:25.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"f81fbb11981cddbd44c7600b6f748148fc80c325","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 18:42:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e0dbb52ddc6a42d3b9acc4768595047a : SUCCESS in 1h 19m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cce15e6cc4f04858b01ab9b39ef5e36a : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/428e0ef02505416fb0e98a89ba9e9ce0 : SUCCESS in 7m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7d18fe26498476a913dfe5680de7d96 : SUCCESS in 6m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2838cf9bda314140b2e76e60a0a9ce44 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71f5c5c08aac4bba961605154c3e7059 : SUCCESS in 7m 18s","accounts_in_message":[],"_revision_number":5},{"id":"b0e146bec630b53103dbc3c15f66e114ac061512","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-11 12:47:10.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"be72394e19248caac84da39e6b43a52e91521adb","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2021-08-24 13:06:05.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"c2e7b1a162c28cc1976e68b0f0b7e56736b46b19","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-11-12 10:20:56.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"c9e66cd8ef4168252f73fda14ecf85358ad6a091","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 10:22:56.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"1d0d0be7288a46cf702186378832262db0a84104","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 12:20:53.000000000","message":"Patch Set 5: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"ff4a9a3c38e4e5514ff0410c8c2c99c702c9a452","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-11-15 11:29:39.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"732f0120ebb6eb9ed77c2e2df32e3b0929f83176","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 13:24:30.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ffe856ba6b114999aabfacf6f25f7eb4 : SUCCESS in 1h 45m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dc05d6292724d30b4fe82c8193a38f6 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28468cdb297c4459ab487fc039710a56 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83bb1658dc17419ab0c9919a234c649d : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e65a9f5d2ee40edbe6430e8eb0acebf : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/984e209252274664824767868c5df932 : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":5},{"id":"a9fb0d5a7af70bff60d9be49a54432217c90da14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 13:25:24.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"12ea84103a54c68253fbfecfbaadb1349c28a2c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 14:45:17.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/18300ebbe4e3475d9fbdbc91d7070aab : SUCCESS in 1h 18m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49c37d46b2b5484ea2857de27e5a52e5 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f9a1b82f5214ee1aa583b8f6db44d07 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3fbe67553da04e55972542f4912e7b80 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3ac2b9ad19b4932a2e692334d4b6b86 : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":5},{"id":"282575a71c1742d93fcbbff8926752fc6a7b30b8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 14:45:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"50d46bf943390b0518056c2de1b1cd7aeaa416f8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 14:47:31.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b42e79a74cfa473dad44ab83e4ca8687 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"ecb15c4ac63772e3332ac0f475e852f414ca3bbb","revisions":{"86860c4607c7f952760d54272b8ea3d4f325c317":{"kind":"REWORK","_number":1,"created":"2021-07-27 17:30:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/85/802585/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/85/802585/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/85/802585/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/85/802585/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/85/802585/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/85/802585/1"}}},"commit":{"parents":[{"commit":"9177ba989412fcb44d3db519c6a27300195d5f57","subject":"Don\u0027t pass kwargs to connection.execute()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/9177ba989412fcb44d3db519c6a27300195d5f57"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-19 10:40:08.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-19 16:44:17.000000000","tz":60},"subject":"Don\u0027t call mapper() outside of declarative registry","message":"Don\u0027t call mapper() outside of declarative registry\n\nResolve the following RemovedIn20Warning:\n\n  Calling the mapper() function directly outside of a declarative\n  registry is deprecated. Please use the\n  sqlalchemy.orm.registry.map_imperatively() function for a classical\n  mapping.\n\nChange-Id: I92a7ccdd48eedd4c788384033743daf50a9dc113\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/86860c4607c7f952760d54272b8ea3d4f325c317"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/86860c4607c7f952760d54272b8ea3d4f325c317"}]},"branch":"refs/heads/master"},"33ec04e99562818a8bc6ab634f9a95fe61125fc1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-07-28 14:38:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/85/802585/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/85/802585/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/85/802585/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/85/802585/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/85/802585/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/85/802585/2"}}},"commit":{"parents":[{"commit":"371dea3368bb5c738490532afc817d69e13b8439","subject":"Don\u0027t pass kwargs to connection.execute()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/371dea3368bb5c738490532afc817d69e13b8439"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-19 10:40:08.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-28 14:38:06.000000000","tz":60},"subject":"Don\u0027t call mapper() outside of declarative registry","message":"Don\u0027t call mapper() outside of declarative registry\n\nResolve the following RemovedIn20Warning:\n\n  Calling the mapper() function directly outside of a declarative\n  registry is deprecated. Please use the\n  sqlalchemy.orm.registry.map_imperatively() function for a classical\n  mapping.\n\nChange-Id: I92a7ccdd48eedd4c788384033743daf50a9dc113\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/33ec04e99562818a8bc6ab634f9a95fe61125fc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/33ec04e99562818a8bc6ab634f9a95fe61125fc1"}]},"branch":"refs/heads/master"},"7558498b96c896e1af879c1fd2b276cbfa84ffc5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-07-29 10:44:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/85/802585/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/85/802585/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/85/802585/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/85/802585/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/85/802585/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/85/802585/3"}}},"commit":{"parents":[{"commit":"b928ed659fab428fb21ffc0d2c55e598a0bea1c1","subject":"Don\u0027t pass kwargs to connection.execute()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/b928ed659fab428fb21ffc0d2c55e598a0bea1c1"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-19 10:40:08.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-29 10:23:07.000000000","tz":60},"subject":"Don\u0027t call mapper() outside of declarative registry","message":"Don\u0027t call mapper() outside of declarative registry\n\nResolve the following RemovedIn20Warning:\n\n  Calling the mapper() function directly outside of a declarative\n  registry is deprecated. Please use the\n  sqlalchemy.orm.registry.map_imperatively() function for a classical\n  mapping.\n\nChange-Id: I92a7ccdd48eedd4c788384033743daf50a9dc113\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7558498b96c896e1af879c1fd2b276cbfa84ffc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7558498b96c896e1af879c1fd2b276cbfa84ffc5"}]},"branch":"refs/heads/master"},"d59f3bc21481e5d5bfb9647374dda16d8a612a90":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-07-29 15:37:53.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/85/802585/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/85/802585/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/85/802585/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/85/802585/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/85/802585/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/85/802585/4"}}},"commit":{"parents":[{"commit":"33f67199f11bac21261a7549a1b593f4d2e2341b","subject":"Don\u0027t pass kwargs to connection.execute()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/33f67199f11bac21261a7549a1b593f4d2e2341b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-19 10:40:08.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-29 15:37:46.000000000","tz":60},"subject":"Don\u0027t call mapper() outside of declarative registry","message":"Don\u0027t call mapper() outside of declarative registry\n\nResolve the following RemovedIn20Warning:\n\n  Calling the mapper() function directly outside of a declarative\n  registry is deprecated. Please use the\n  sqlalchemy.orm.registry.map_imperatively() function for a classical\n  mapping.\n\nChange-Id: I92a7ccdd48eedd4c788384033743daf50a9dc113\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/d59f3bc21481e5d5bfb9647374dda16d8a612a90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/d59f3bc21481e5d5bfb9647374dda16d8a612a90"}]},"branch":"refs/heads/master"},"ecb15c4ac63772e3332ac0f475e852f414ca3bbb":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-08-10 16:45:25.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/85/802585/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/85/802585/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/85/802585/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/85/802585/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/85/802585/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/85/802585/5"}}},"commit":{"parents":[{"commit":"e1039e0849890c6bd10516004c6dab89db1c9111","subject":"Don\u0027t pass kwargs to connection.execute()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e1039e0849890c6bd10516004c6dab89db1c9111"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-19 10:40:08.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-08-10 16:38:16.000000000","tz":60},"subject":"Don\u0027t call mapper() outside of declarative registry","message":"Don\u0027t call mapper() outside of declarative registry\n\nResolve the following RemovedIn20Warning:\n\n  Calling the mapper() function directly outside of a declarative\n  registry is deprecated. Please use the\n  sqlalchemy.orm.registry.map_imperatively() function for a classical\n  mapping.\n\nChange-Id: I92a7ccdd48eedd4c788384033743daf50a9dc113\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/ecb15c4ac63772e3332ac0f475e852f414ca3bbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/ecb15c4ac63772e3332ac0f475e852f414ca3bbb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
