)]}'
{"id":"openstack%2Foslo.db~802586","triplet_id":"openstack%2Foslo.db~master~I4c47a690a94abcb3b4b6fb087a1bf86c5350b523","project":"openstack/oslo.db","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2021-11-15 11:29:44.000000000","reason":"stephenfin replied on the change"}},"hashtags":[],"change_id":"I4c47a690a94abcb3b4b6fb087a1bf86c5350b523","subject":"Replace use of \u0027Engine.execute()\u0027","status":"MERGED","created":"2021-07-27 17:30:19.000000000","updated":"2021-11-15 15:02:31.000000000","submitted":"2021-11-15 14:59:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"802586-sqlalchemy-20","meta_rev_id":"e6e10f1b671578c7610d54b10c7bbc675cb06376","_number":802586,"virtual_id_number":802586,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-15 14:59:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2021-08-24 13:08:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":2,"date":"2021-08-11 12:51:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-11-12 10:21:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-27 19:14:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-11 12:51:28.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-08-24 13:08:53.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"00a212b3c9c47684d8423ba4875424648aea87e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-27 17:30:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8bb43ea9e3b43116777f2ca2ecc1e2b0f83de509","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 19:14:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/29b1e3bf7fd04c7fad956b0b07b4811a : FAILURE in 1h 07m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f82f83a77b454d499a4e4411ab1f7a61 : FAILURE in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f83cb1aeca274ba0a1e747a76064b870 : FAILURE in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb3a84db3bb94fb692794f3d2c46e3e1 : FAILURE in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dee28e5708654a7fb57d998990b44fbc : FAILURE in 7m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cda9c40c99e546cdb0f14e7202f398f4 : SUCCESS in 7m 03s\n\nWarning:\n  Comments left for invalid file oslo_db/sqlalchemy/engines.py","accounts_in_message":[],"_revision_number":1},{"id":"0167854e5134a3b27890e406b90dff419912419c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-28 14:38:18.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"8f02ec00b315c6b0ded5856c90aa02f93575b136","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 16:07:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/62b136173c514f1da44f798fec729a60 : SUCCESS in 1h 20m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a08c39d543704b3ea5a48809ac834117 : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ac98f27d34e44cea1a1cc6ecf567eb4 : FAILURE in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d41650ccc6a4af9ab90121dd9a29730 : FAILURE in 7m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b607004e85ea440581db1e507ed18c77 : FAILURE in 7m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e3082206261493282fee79f0ea7b5d8 : SUCCESS in 7m 52s","accounts_in_message":[],"_revision_number":2},{"id":"824cf09c23d3ae84762ca12257ac4e281abbf5f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-29 10:44:29.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"9f2c8e2b2ca91f748e95361c2335b83bbc22b0ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 12:24:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c97f59a561e545d9aadb8e77a5ee1ca0 : SUCCESS in 1h 28m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4845db70f974b8390cc43eb1d6e3aa9 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0999bba4174e4ded9aef2fc575eab82c : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4829598762034a058038ca23f770fd64 : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65691d1c83d44b868e155dabb161d107 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6277f54521cb48d7b377901bce64f1df : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":3},{"id":"64c0bc8192f09c7fa1f07d05722da882fc209b55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-29 15:37:53.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"282b2ba66cce137cfb6ce8b92f48d76b53eb34a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 17:06:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/25f1f30a0e5c4a8db194b2d5b239ce64 : SUCCESS in 1h 21m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e75b15f0560745edbc956ed22f9a8bab : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40198d562656431da94dbdb6a3351ad2 : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/273e61d9978446ee96df8abcd119cdcb : SUCCESS in 8m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f4f0e657f0c4256b9fa6981ee90da95 : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7db804e05cf24988997b9a9f26f015b8 : SUCCESS in 8m 59s","accounts_in_message":[],"_revision_number":4},{"id":"be7ba265f021cd6779fcb95f5bf0ccd63f5e3081","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-10 16:45:25.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"a7fdcd499ace7d81e99dd9971268995f272334d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 18:48:43.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa58aabeba1a4182b89972cdd3ebaa92 : SUCCESS in 1h 11m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaf2363a3dbf4c9ebf8faca9c699a74e : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e29c20e12464f9a87a037b5ca108b2f : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15219ad50ec14b699188d154cd18d3be : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60047e98d0214376b68000d47230ac29 : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e01daae036644be284d71d1734895156 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":5},{"id":"6079befb7b01ce9f7a6a5b3334ba7d06a09fa77b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-11 12:51:28.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"140c5620750590e758e06ab8d5e03d3a6eb30859","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2021-08-24 13:08:53.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)\n\nconnectable.connect() will have to change but that\u0027s a separate deprecation warning","accounts_in_message":[],"_revision_number":5},{"id":"1fb377b56f3758dc51e76c20219583b69decda8e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-25 10:20:00.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"34ea87b2fee9af9c9185f3e67536b29930a7c62a","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2021-08-25 13:52:55.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"072034ad78b870e1aa3e20b6de0587d569890699","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-11-12 10:21:02.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"878f7b01fd231478c2ec9aaff5ce9e2e8c89c75d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 10:22:57.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"cc8197599a07fc19e09af8188c699781230f207b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 12:20:54.000000000","message":"Patch Set 5: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"a2f68c1a1265797c84ccb1df1cfb6a4fc5b7bdfc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-11-15 11:29:44.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f583f089be1e661262e978a3ab08f41547f61c10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 13:09:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5f7404f617ff448eb02594a50e141637 : SUCCESS in 1h 32m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/825108be20b244878677bc12e5b69b56 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02092726d6b842a79cab069239f8375e : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00c6aa13c5d24460a4d6fe1d31e4c294 : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c3254b080c1471e86dcbebbf4417705 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd88cb883350446894b8cfdb5cd24359 : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":5},{"id":"7c25c1a66c32b7ce139e565fa1690d82755de701","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 13:25:25.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"665824cc352b9dbd935060a3c9527da47e880e8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 14:59:26.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/256618a48b1545fc857765293a5b3538 : SUCCESS in 1h 33m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2831caa6f684797acf043933c2fd675 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5514b19d968e42789f1cb7e2eb50a5c7 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/940040df935a4a1ca2c73025d34d6d97 : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8ca1a72e5bd4d22a0f78e620138ae2d : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":5},{"id":"0bd8159599c7a028926a039976fed0b418fc06ac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 14:59:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"e6e10f1b671578c7610d54b10c7bbc675cb06376","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 15:02:31.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d27e15e0c2b424a87ca378ca4a6aee4 : SUCCESS in 1m 57s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"df901a1c765c16de77c2f734760e9ecf95483737","revisions":{"9b4700aaab0b110433f7ef32146b48c1581449f3":{"kind":"REWORK","_number":1,"created":"2021-07-27 17:30:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/86/802586/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/86/802586/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/86/802586/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/86/802586/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/86/802586/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/86/802586/1"}}},"commit":{"parents":[{"commit":"86860c4607c7f952760d54272b8ea3d4f325c317","subject":"Don\u0027t call mapper() outside of declarative registry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/86860c4607c7f952760d54272b8ea3d4f325c317"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-19 10:48:58.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-19 16:44:17.000000000","tz":60},"subject":"Replace use of \u0027Engine.execute()\u0027","message":"Replace use of \u0027Engine.execute()\u0027\n\nResolve the following RemovedIn20Warning warning:\n\n  The Engine.execute() method is considered legacy as of the 1.x series\n  of SQLAlchemy and will be removed in 2.0. All statement execution in\n  SQLAlchemy 2.0 is performed by the Connection.execute() method of\n  Connection, or in the ORM by the Session.execute() method of Session.\n\nChange-Id: I4c47a690a94abcb3b4b6fb087a1bf86c5350b523\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/9b4700aaab0b110433f7ef32146b48c1581449f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/9b4700aaab0b110433f7ef32146b48c1581449f3"}]},"branch":"refs/heads/master"},"593d6d10e983ea0bbf429f67b3ba3da64116001e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-07-28 14:38:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/86/802586/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/86/802586/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/86/802586/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/86/802586/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/86/802586/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/86/802586/2"}}},"commit":{"parents":[{"commit":"33ec04e99562818a8bc6ab634f9a95fe61125fc1","subject":"Don\u0027t call mapper() outside of declarative registry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/33ec04e99562818a8bc6ab634f9a95fe61125fc1"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-19 10:48:58.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-28 14:38:06.000000000","tz":60},"subject":"Replace use of \u0027Engine.execute()\u0027","message":"Replace use of \u0027Engine.execute()\u0027\n\nResolve the following RemovedIn20Warning warning:\n\n  The Engine.execute() method is considered legacy as of the 1.x series\n  of SQLAlchemy and will be removed in 2.0. All statement execution in\n  SQLAlchemy 2.0 is performed by the Connection.execute() method of\n  Connection, or in the ORM by the Session.execute() method of Session.\n\nChange-Id: I4c47a690a94abcb3b4b6fb087a1bf86c5350b523\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/593d6d10e983ea0bbf429f67b3ba3da64116001e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/593d6d10e983ea0bbf429f67b3ba3da64116001e"}]},"branch":"refs/heads/master"},"de636e9b78b0234bc1481af89b5920ea417f38ef":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-07-29 10:44:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/86/802586/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/86/802586/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/86/802586/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/86/802586/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/86/802586/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/86/802586/3"}}},"commit":{"parents":[{"commit":"7558498b96c896e1af879c1fd2b276cbfa84ffc5","subject":"Don\u0027t call mapper() outside of declarative registry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7558498b96c896e1af879c1fd2b276cbfa84ffc5"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-19 10:48:58.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-29 10:23:07.000000000","tz":60},"subject":"Replace use of \u0027Engine.execute()\u0027","message":"Replace use of \u0027Engine.execute()\u0027\n\nResolve the following RemovedIn20Warning warning:\n\n  The Engine.execute() method is considered legacy as of the 1.x series\n  of SQLAlchemy and will be removed in 2.0. All statement execution in\n  SQLAlchemy 2.0 is performed by the Connection.execute() method of\n  Connection, or in the ORM by the Session.execute() method of Session.\n\nChange-Id: I4c47a690a94abcb3b4b6fb087a1bf86c5350b523\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/de636e9b78b0234bc1481af89b5920ea417f38ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/de636e9b78b0234bc1481af89b5920ea417f38ef"}]},"branch":"refs/heads/master"},"e19c47e336ea1925983f6f941498ce3b703252b8":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-07-29 15:37:53.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/86/802586/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/86/802586/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/86/802586/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/86/802586/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/86/802586/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/86/802586/4"}}},"commit":{"parents":[{"commit":"d59f3bc21481e5d5bfb9647374dda16d8a612a90","subject":"Don\u0027t call mapper() outside of declarative registry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/d59f3bc21481e5d5bfb9647374dda16d8a612a90"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-19 10:48:58.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-29 15:37:47.000000000","tz":60},"subject":"Replace use of \u0027Engine.execute()\u0027","message":"Replace use of \u0027Engine.execute()\u0027\n\nResolve the following RemovedIn20Warning warning:\n\n  The Engine.execute() method is considered legacy as of the 1.x series\n  of SQLAlchemy and will be removed in 2.0. All statement execution in\n  SQLAlchemy 2.0 is performed by the Connection.execute() method of\n  Connection, or in the ORM by the Session.execute() method of Session.\n\nChange-Id: I4c47a690a94abcb3b4b6fb087a1bf86c5350b523\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e19c47e336ea1925983f6f941498ce3b703252b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e19c47e336ea1925983f6f941498ce3b703252b8"}]},"branch":"refs/heads/master"},"df901a1c765c16de77c2f734760e9ecf95483737":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-08-10 16:45:25.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/86/802586/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/86/802586/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/86/802586/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/86/802586/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/86/802586/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/86/802586/5"}}},"commit":{"parents":[{"commit":"ecb15c4ac63772e3332ac0f475e852f414ca3bbb","subject":"Don\u0027t call mapper() outside of declarative registry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/ecb15c4ac63772e3332ac0f475e852f414ca3bbb"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-19 10:48:58.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-08-10 16:38:48.000000000","tz":60},"subject":"Replace use of \u0027Engine.execute()\u0027","message":"Replace use of \u0027Engine.execute()\u0027\n\nResolve the following RemovedIn20Warning warning:\n\n  The Engine.execute() method is considered legacy as of the 1.x series\n  of SQLAlchemy and will be removed in 2.0. All statement execution in\n  SQLAlchemy 2.0 is performed by the Connection.execute() method of\n  Connection, or in the ORM by the Session.execute() method of Session.\n\nChange-Id: I4c47a690a94abcb3b4b6fb087a1bf86c5350b523\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/df901a1c765c16de77c2f734760e9ecf95483737"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/df901a1c765c16de77c2f734760e9ecf95483737"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
