)]}'
{"id":"openstack%2Foslo.db~802587","triplet_id":"openstack%2Foslo.db~master~Ib789cd4d11a3d5dd01fcdb99822025b11bbc234e","project":"openstack/oslo.db","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-02-02 21:00:32.000000000","reason":"Change was submitted"},"11816":{"account":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"last_update":"2022-02-02 19:27:56.000000000","reason":"mike_mp@zzzcomputing.com replied on the change"}},"hashtags":[],"change_id":"Ib789cd4d11a3d5dd01fcdb99822025b11bbc234e","subject":"Don\u0027t rely on implicit autocommit","status":"MERGED","created":"2021-07-27 17:30:19.000000000","updated":"2022-02-02 21:03:23.000000000","submitted":"2022-02-02 21:00:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802587-sqlalchemy-20","meta_rev_id":"65073e8bbfe1dd6f281cf0e1f140e97a68d225a2","_number":802587,"virtual_id_number":802587,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-02 21:00:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":2,"date":"2022-02-02 19:27:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":2,"date":"2021-08-11 13:01:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":1,"date":"2022-02-02 19:27:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-27 19:13:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-11 13:01:22.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-11-12 10:21:15.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"de2c14bac3afc3ab65acc7b339336e265e1fedb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-27 17:30:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c85050f466935dad30ccea45a63c9d975fe85a27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 19:13:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b34e6e875a7b43da89738fdcbf6e1612 : FAILURE in 1h 01m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/240a28a7dbb3422bb3314bece4427048 : FAILURE in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab066833b881478b963b55d676e905ea : FAILURE in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b1c78ee669d453a9bcdce41313775c5 : FAILURE in 7m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d50c2ac502d9492ebcf022373e3224ed : FAILURE in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/009aa97e9aee4ef98f4b3b8d45549512 : SUCCESS in 6m 03s\n\nWarning:\n  Comments left for invalid file oslo_db/sqlalchemy/engines.py","accounts_in_message":[],"_revision_number":1},{"id":"e624d12db3df9696b2d4299ff4da95962527e025","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-28 14:38:18.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"9ab368934dec839ce247373a35110455f1e4058a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 15:57:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4bd8bddc754d4e9dbd4108ee2f3c3ccf : SUCCESS in 1h 13m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1d3777f09784837936f6952f5db7895 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c794b7adf6e0436b8bfc4b2cd584fb31 : FAILURE in 7m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21248f4f9dd34fd584795efa53e4ef40 : FAILURE in 7m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a67c272bd45b427cbd420c4fe1824b9b : FAILURE in 7m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c48aecc8e1f4a3b82d6a2a79ec3d895 : SUCCESS in 7m 06s","accounts_in_message":[],"_revision_number":2},{"id":"aae8195f1c2345b095a23f6bc717250591e8950f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-29 10:44:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"19c6df16d6c5a6d4d049b7d9d5bf33242dd1a7c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 12:26:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a4c5e650b5cd4d7a9a11a7f3b5befbf3 : SUCCESS in 1h 27m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9612f44f282b43b6954d103a92c70d0a : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c540ebc307ae47e4b84135bd742c8c92 : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1761f2ce5c94fabbe0de8e836b1eb01 : SUCCESS in 8m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e042e65a04c2439785e323de14a49136 : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c786d9a6ee0947cbbd5385f83a8eee15 : SUCCESS in 6m 59s","accounts_in_message":[],"_revision_number":3},{"id":"f7d6b8e319437894732031013e804b157c7fb9fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-29 15:37:53.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"9711a90a5361f6daf0eb7a0d569d01ebc21e09ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 17:18:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d46892e938ae4a9a84bbfe789081484b : SUCCESS in 1h 30m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff89da55fd074e66b9a1a0cb7955d32c : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d4ada04204f4367b1c6cde68e905258 : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd7d00ced5084c90be2be23a481c73ed : SUCCESS in 7m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91d3dad41c404a4a890f2d4d90b5635e : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77322e921c074e88a47a09c97b20b59b : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":4},{"id":"b566192a5148257aa55321eff5fbcd130db3a833","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-10 16:45:25.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"dd9da62d9d3b9dc7c14e5c46afd5133c7ae5ebb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 19:15:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e6e7f5a133574c4c800f2c49c45bd4ca : SUCCESS in 1h 33m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/647d42cec09b4dcba6f898392401f4c2 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0946cf28c6a44b881c5540446b74f83 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d2fa90e8acb439bac0561091bf3ec1d : SUCCESS in 7m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/259b64c4282f42349963869a91f5a34d : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03c120dbc5b54885bb4c6e3873c36f58 : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":5},{"id":"d30870e75ba52afac05ad9e34a37a6d4adec5d79","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-11 13:01:22.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"1bff992ca65e64b93044fafd816a15a4bcb33a61","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2022-02-02 19:27:56.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"6fa6b3068409a59b8e649c3d01a9b272d67ec93c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 19:28:57.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"035cda4dcfdd04e42574f6e19689850c0a3bd8a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 21:00:32.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dcaf29d4a19940cc8926c6b996b9a870 : SUCCESS in 1h 27m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b6e2eb4280c484885883948c67084fe : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3843c96905f04ed3a612cad9b3aabb9b : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96cbfa6a105c46059b5f2b348005dddf : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/994f9f4ca4b64ee2ab611eecb179a856 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":5},{"id":"7b7d4205b49a9ec404972bdea36b018f9daf81e7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 21:00:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"65073e8bbfe1dd6f281cf0e1f140e97a68d225a2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 21:03:23.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9f84d57e21a445eb4d4d50b3a6f8767 : SUCCESS in 2m 13s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"b3a56b35640b0cfceeeb97c885318a3aae7f7b22","revisions":{"42470a0ce3fb53fd4815d271719944a8471b69e8":{"kind":"REWORK","_number":1,"created":"2021-07-27 17:30:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/87/802587/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/87/802587/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/87/802587/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/87/802587/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/87/802587/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/87/802587/1"}}},"commit":{"parents":[{"commit":"9b4700aaab0b110433f7ef32146b48c1581449f3","subject":"Replace use of \u0027Engine.execute()\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/9b4700aaab0b110433f7ef32146b48c1581449f3"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 15:55:56.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-19 16:44:17.000000000","tz":60},"subject":"Don\u0027t rely on implicit autocommit","message":"Don\u0027t rely on implicit autocommit\n\nResolve the following RemovedIn20Warning warning:\n\n  The current statement is being autocommitted using implicit\n  autocommit, which will be removed in SQLAlchemy 2.0. Use the .begin()\n  method of Engine or Connection in order to use an explicit transaction\n  for DML and DDL statements.\n\nChange-Id: Ib789cd4d11a3d5dd01fcdb99822025b11bbc234e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/42470a0ce3fb53fd4815d271719944a8471b69e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/42470a0ce3fb53fd4815d271719944a8471b69e8"}]},"branch":"refs/heads/master"},"502177374ee0a5693337321211ae163371858604":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-07-28 14:38:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/87/802587/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/87/802587/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/87/802587/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/87/802587/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/87/802587/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/87/802587/2"}}},"commit":{"parents":[{"commit":"593d6d10e983ea0bbf429f67b3ba3da64116001e","subject":"Replace use of \u0027Engine.execute()\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/593d6d10e983ea0bbf429f67b3ba3da64116001e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 15:55:56.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-28 14:38:06.000000000","tz":60},"subject":"Don\u0027t rely on implicit autocommit","message":"Don\u0027t rely on implicit autocommit\n\nResolve the following RemovedIn20Warning warning:\n\n  The current statement is being autocommitted using implicit\n  autocommit, which will be removed in SQLAlchemy 2.0. Use the .begin()\n  method of Engine or Connection in order to use an explicit transaction\n  for DML and DDL statements.\n\nChange-Id: Ib789cd4d11a3d5dd01fcdb99822025b11bbc234e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/502177374ee0a5693337321211ae163371858604"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/502177374ee0a5693337321211ae163371858604"}]},"branch":"refs/heads/master"},"041627f9f8220fd0c2f091a911d91830b39e332b":{"kind":"REWORK","_number":3,"created":"2021-07-29 10:44:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/87/802587/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/87/802587/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/87/802587/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/87/802587/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/87/802587/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/87/802587/3"}}},"commit":{"parents":[{"commit":"de636e9b78b0234bc1481af89b5920ea417f38ef","subject":"Replace use of \u0027Engine.execute()\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/de636e9b78b0234bc1481af89b5920ea417f38ef"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 15:55:56.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-29 10:42:29.000000000","tz":60},"subject":"Don\u0027t rely on implicit autocommit","message":"Don\u0027t rely on implicit autocommit\n\nResolve the following RemovedIn20Warning warning:\n\n  The current statement is being autocommitted using implicit\n  autocommit, which will be removed in SQLAlchemy 2.0. Use the .begin()\n  method of Engine or Connection in order to use an explicit transaction\n  for DML and DDL statements.\n\nChange-Id: Ib789cd4d11a3d5dd01fcdb99822025b11bbc234e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/041627f9f8220fd0c2f091a911d91830b39e332b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/041627f9f8220fd0c2f091a911d91830b39e332b"}]},"branch":"refs/heads/master"},"4a8fa58cb454042fb3ecdb010d2ad713a1cbe739":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-07-29 15:37:53.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/87/802587/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/87/802587/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/87/802587/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/87/802587/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/87/802587/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/87/802587/4"}}},"commit":{"parents":[{"commit":"e19c47e336ea1925983f6f941498ce3b703252b8","subject":"Replace use of \u0027Engine.execute()\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e19c47e336ea1925983f6f941498ce3b703252b8"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 15:55:56.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-29 15:37:47.000000000","tz":60},"subject":"Don\u0027t rely on implicit autocommit","message":"Don\u0027t rely on implicit autocommit\n\nResolve the following RemovedIn20Warning warning:\n\n  The current statement is being autocommitted using implicit\n  autocommit, which will be removed in SQLAlchemy 2.0. Use the .begin()\n  method of Engine or Connection in order to use an explicit transaction\n  for DML and DDL statements.\n\nChange-Id: Ib789cd4d11a3d5dd01fcdb99822025b11bbc234e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/4a8fa58cb454042fb3ecdb010d2ad713a1cbe739"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/4a8fa58cb454042fb3ecdb010d2ad713a1cbe739"}]},"branch":"refs/heads/master"},"b3a56b35640b0cfceeeb97c885318a3aae7f7b22":{"kind":"REWORK","_number":5,"created":"2021-08-10 16:45:25.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/87/802587/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/87/802587/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/87/802587/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/87/802587/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/87/802587/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/87/802587/5"}}},"commit":{"parents":[{"commit":"df901a1c765c16de77c2f734760e9ecf95483737","subject":"Replace use of \u0027Engine.execute()\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/df901a1c765c16de77c2f734760e9ecf95483737"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-08-10 16:42:47.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-08-10 16:42:47.000000000","tz":60},"subject":"Don\u0027t rely on implicit autocommit","message":"Don\u0027t rely on implicit autocommit\n\nResolve the following RemovedIn20Warning warning:\n\n  The current statement is being autocommitted using implicit\n  autocommit, which will be removed in SQLAlchemy 2.0. Use the .begin()\n  method of Engine or Connection in order to use an explicit transaction\n  for DML and DDL statements.\n\nChange-Id: Ib789cd4d11a3d5dd01fcdb99822025b11bbc234e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/b3a56b35640b0cfceeeb97c885318a3aae7f7b22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/b3a56b35640b0cfceeeb97c885318a3aae7f7b22"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
