)]}'
{"id":"openstack%2Foslo.db~802588","triplet_id":"openstack%2Foslo.db~master~Ib2537bae77861ee60d17a48a72c57b88e043553e","project":"openstack/oslo.db","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-02-02 21:02:51.000000000","reason":"Change was submitted"},"11816":{"account":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"last_update":"2022-02-02 19:35:59.000000000","reason":"mike_mp@zzzcomputing.com replied on the change"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-02-02 21:02:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib2537bae77861ee60d17a48a72c57b88e043553e","subject":"Remove use of Session.begin.subtransactions flag","status":"MERGED","created":"2021-07-27 17:30:19.000000000","updated":"2022-02-02 21:05:04.000000000","submitted":"2022-02-02 21:02:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802588-sqlalchemy-20","meta_rev_id":"6669420c213494bc19ea5f0e566990a7125f8236","_number":802588,"virtual_id_number":802588,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-02 21:02:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":2,"date":"2022-02-02 19:35:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":1,"date":"2021-08-11 12:57:44.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":1,"date":"2022-02-02 19:35:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-27 19:19:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-11 12:57:44.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-11-12 10:21:40.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"3f21b69637f78f8f362db00acbc857606ba479f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-27 17:30:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0ce656c5369d18552d988cb2be493a4e787678f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 19:19:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cfe35c8f1d1e4581aec1ecedcc58a09f : FAILURE in 1h 05m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/874432cc45894ccf89eb4087fddad1e6 : FAILURE in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7cef6b6e43854ee198627bec97a8c2ee : FAILURE in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba4f78b8183c45c385dc33b41645f6e0 : FAILURE in 8m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f6248a72d164dff8cc01272bac2fd03 : FAILURE in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b94ac29dde4b47e78a9d257c23880a46 : SUCCESS in 7m 33s\n\nWarning:\n  Comments left for invalid file oslo_db/sqlalchemy/engines.py","accounts_in_message":[],"_revision_number":1},{"id":"0caf18525c3ab72a81ba4f682378deccb73f1bfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-28 14:38:18.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"5b5f1a34957e9fa738f223f91c2e1aa6405845e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 16:10:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/825a8a8682da4b039e5f6b85f242a43b : SUCCESS in 1h 23m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f78ba8421fb493c8926fc526a005e6e : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9eac9a518e074beb8c7d72d1a65765e1 : FAILURE in 8m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3190d312e9174cf5895683586a9336ce : FAILURE in 7m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/80a5318b25bc4184aff91313ca657f84 : FAILURE in 7m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/978efa31a8bf437b82888df88f6f00d0 : SUCCESS in 7m 52s","accounts_in_message":[],"_revision_number":2},{"id":"02d75bbfd9f91182045c67fb7bf71869df73e58b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-29 10:44:29.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"190c2b50c276c959868b452b9e61600328220ad8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 12:27:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e20dd05c035f4b96acd24e1a15a751f3 : SUCCESS in 1h 28m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a55666b6f58f4d84a24384396ab63588 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40dcaaaac81c47f4bfcc6e4221426ca8 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/926ba8e181624c9fb39d4a87a71afe64 : SUCCESS in 6m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/709b1faee16f407f9e45e8d14c044e21 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12e16f37f4314c49a2b53fd0a2cdc6bf : SUCCESS in 6m 46s","accounts_in_message":[],"_revision_number":3},{"id":"5275f1e1833305605d89bd00c6107bcfc6500bcb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-29 15:37:53.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"e1aed79b6522e08760e86ecde6ef234c837bf6f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 17:21:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ad9297434cb94627b52bf478acbb8b83 : SUCCESS in 1h 33m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13d0fc492bfa4f778d60082a712b1bb7 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d97b6e50b4d64af9995c49873d8f0301 : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/696eedc98b14496286e097deb962144a : SUCCESS in 7m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c0c774e2f68476283a00612ab178e8c : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/050e78cad8ba47f8aadbac1fda0f1805 : SUCCESS in 7m 34s","accounts_in_message":[],"_revision_number":4},{"id":"63eaf5cfb9d289f8017d78cf4f4cdfcb77aec479","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-10 16:45:25.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"a2287baa8dc50e79dba163dbbd271feda431bbde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 19:05:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2cafa329d72249d0935f02e4b7c2092e : SUCCESS in 1h 17m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86b339c74aa24aca96a8d6912d027cfb : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1be4a752c124d5394290f1bbd0d671b : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/519a1fec5ff84147868840aecd608b96 : SUCCESS in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe6f24bd94fe43b1bcc6e30da440a58b : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ce9791453584d3496f44bf1ef4ef157 : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":5},{"id":"48e24c867ded7c08547a1fbef3a6462f7d1bea09","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-11 12:57:44.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ebd469d8918b8e9c1dd942f89cb3b8e790a09c76","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-19 09:43:21.000000000","message":"Patch Set 5:\n\nIf Mike think that my previous concern isn\u0027t an issue then feel free to consider my +1 as a +2","accounts_in_message":[],"_revision_number":5},{"id":"e92859bf5e6427a229f3f1092be54ea88a033f2b","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2022-02-02 19:35:59.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"eace847a66ba80f0a8fb41de6d63bb23b1b84a78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 19:37:35.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"38e2f35d9309bb43a3bd7da81dd122f20122ea43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 21:02:51.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b5a7d86f2eea41fb9d0db439bd0e3169 : SUCCESS in 1h 23m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1277a82c4e1f4c92ab0ce816ce7b1406 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a24468a4d20b4b8bb0ca426fe2b6b447 : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab7668094fb54f53b71b04499d40c24a : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dba45f9a2a240fca7fb9c90aa6d7bf9 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":5},{"id":"70ebdcdc5edbe6c4f537b2a3e7e36929d0d293f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 21:02:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"6669420c213494bc19ea5f0e566990a7125f8236","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 21:05:04.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b195cb590934ace86ff186d995bebec : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"cd5f697a173a64df0cdb1a90ab871b81f56469d4","revisions":{"fbc26d5e6b4436d7c483123cf2488a19255e95ba":{"kind":"REWORK","_number":1,"created":"2021-07-27 17:30:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/802588/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/88/802588/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/88/802588/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/88/802588/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/88/802588/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/88/802588/1"}}},"commit":{"parents":[{"commit":"42470a0ce3fb53fd4815d271719944a8471b69e8","subject":"Don\u0027t rely on implicit autocommit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/42470a0ce3fb53fd4815d271719944a8471b69e8"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 14:41:05.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-19 16:44:17.000000000","tz":60},"subject":"Remove use of Session.begin.subtransactions flag","message":"Remove use of Session.begin.subtransactions flag\n\nResolve the following RemovedIn20Warning warning:\n\n  The Session.begin.subtransactions flag is deprecated and will be\n  removed in SQLAlchemy version 2.0.  See the documentation at\n  session_subtransactions for background on a compatible alternative\n  pattern.\n\nChange-Id: Ib2537bae77861ee60d17a48a72c57b88e043553e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/fbc26d5e6b4436d7c483123cf2488a19255e95ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/fbc26d5e6b4436d7c483123cf2488a19255e95ba"}]},"branch":"refs/heads/master"},"266210df5dc42e0a732b209d8bdc2719525af2fc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-07-28 14:38:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/802588/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/88/802588/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/88/802588/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/88/802588/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/88/802588/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/88/802588/2"}}},"commit":{"parents":[{"commit":"502177374ee0a5693337321211ae163371858604","subject":"Don\u0027t rely on implicit autocommit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/502177374ee0a5693337321211ae163371858604"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 14:41:05.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-28 14:38:06.000000000","tz":60},"subject":"Remove use of Session.begin.subtransactions flag","message":"Remove use of Session.begin.subtransactions flag\n\nResolve the following RemovedIn20Warning warning:\n\n  The Session.begin.subtransactions flag is deprecated and will be\n  removed in SQLAlchemy version 2.0.  See the documentation at\n  session_subtransactions for background on a compatible alternative\n  pattern.\n\nChange-Id: Ib2537bae77861ee60d17a48a72c57b88e043553e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/266210df5dc42e0a732b209d8bdc2719525af2fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/266210df5dc42e0a732b209d8bdc2719525af2fc"}]},"branch":"refs/heads/master"},"0b2f1c056e42c9a29906d0565d0a91c3b2c65f81":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-07-29 10:44:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/802588/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/88/802588/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/88/802588/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/88/802588/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/88/802588/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/88/802588/3"}}},"commit":{"parents":[{"commit":"041627f9f8220fd0c2f091a911d91830b39e332b","subject":"Don\u0027t rely on implicit autocommit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/041627f9f8220fd0c2f091a911d91830b39e332b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 14:41:05.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-29 10:44:19.000000000","tz":60},"subject":"Remove use of Session.begin.subtransactions flag","message":"Remove use of Session.begin.subtransactions flag\n\nResolve the following RemovedIn20Warning warning:\n\n  The Session.begin.subtransactions flag is deprecated and will be\n  removed in SQLAlchemy version 2.0.  See the documentation at\n  session_subtransactions for background on a compatible alternative\n  pattern.\n\nChange-Id: Ib2537bae77861ee60d17a48a72c57b88e043553e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/0b2f1c056e42c9a29906d0565d0a91c3b2c65f81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/0b2f1c056e42c9a29906d0565d0a91c3b2c65f81"}]},"branch":"refs/heads/master"},"127b50cd7557dbde63c6354262024e4e89929f43":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-07-29 15:37:53.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/802588/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/88/802588/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/88/802588/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/88/802588/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/88/802588/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/88/802588/4"}}},"commit":{"parents":[{"commit":"4a8fa58cb454042fb3ecdb010d2ad713a1cbe739","subject":"Don\u0027t rely on implicit autocommit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/4a8fa58cb454042fb3ecdb010d2ad713a1cbe739"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 14:41:05.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-29 15:37:47.000000000","tz":60},"subject":"Remove use of Session.begin.subtransactions flag","message":"Remove use of Session.begin.subtransactions flag\n\nResolve the following RemovedIn20Warning warning:\n\n  The Session.begin.subtransactions flag is deprecated and will be\n  removed in SQLAlchemy version 2.0.  See the documentation at\n  session_subtransactions for background on a compatible alternative\n  pattern.\n\nChange-Id: Ib2537bae77861ee60d17a48a72c57b88e043553e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/127b50cd7557dbde63c6354262024e4e89929f43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/127b50cd7557dbde63c6354262024e4e89929f43"}]},"branch":"refs/heads/master"},"cd5f697a173a64df0cdb1a90ab871b81f56469d4":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-08-10 16:45:25.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/802588/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/88/802588/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/88/802588/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/88/802588/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/88/802588/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/88/802588/5"}}},"commit":{"parents":[{"commit":"b3a56b35640b0cfceeeb97c885318a3aae7f7b22","subject":"Don\u0027t rely on implicit autocommit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/b3a56b35640b0cfceeeb97c885318a3aae7f7b22"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-16 14:41:05.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-08-10 16:43:22.000000000","tz":60},"subject":"Remove use of Session.begin.subtransactions flag","message":"Remove use of Session.begin.subtransactions flag\n\nResolve the following RemovedIn20Warning warning:\n\n  The Session.begin.subtransactions flag is deprecated and will be\n  removed in SQLAlchemy version 2.0.  See the documentation at\n  session_subtransactions for background on a compatible alternative\n  pattern.\n\nChange-Id: Ib2537bae77861ee60d17a48a72c57b88e043553e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/cd5f697a173a64df0cdb1a90ab871b81f56469d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/cd5f697a173a64df0cdb1a90ab871b81f56469d4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
