)]}'
{"id":"openstack%2Foslo.db~802731","triplet_id":"openstack%2Foslo.db~stable%2Ftrain~I07b577f4e948c03d1751145e5f898b90ac1c197a","project":"openstack/oslo.db","branch":"stable/train","topic":"oslo_lc_drop","hashtags":[],"change_id":"I07b577f4e948c03d1751145e5f898b90ac1c197a","subject":"Remove lower-constraints remnants","status":"MERGED","created":"2021-07-28 14:47:50.000000000","updated":"2021-08-06 11:15:25.000000000","submitted":"2021-08-06 11:13:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802731-oslo_lc_drop","meta_rev_id":"7e5ba3581ec97efb98329be243a1cac80b7489b1","_number":802731,"virtual_id_number":802731,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-06 11:13:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-08-06 09:47:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-08-06 09:47:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"}],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-28 14:47:50.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2021-07-28 16:35:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-06 07:47:13.000000000","updated_by":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"reviewer":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"state":"REVIEWER"}],"messages":[{"id":"b66d25b610addcc29c4e30b0218efb638a3d76f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-28 14:47:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9c9efec4997816d9ede3c372ccee4e83dca6f4ba","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-28 14:48:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3750fcdc2c0f648e9606519b3d001a36c287f635","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 16:35:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7905ee2757a945eb92a750eb984cddaf : RETRY_LIMIT in 4m 14s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d2a98410d4394c348c448a249b7cc804 : SUCCESS in 1h 21m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/546f1017ee1a42eaaa9a55b8d9e5e7d3 : SUCCESS in 1h 43m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6972094f7c3c47608120f943a4adb15e : SUCCESS in 6m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ab261fa509f24de28b06b085d84de6a0 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5dd7c6f0ed404effba1f6497c994424e : SUCCESS in 6m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2cefc88095164bc88d5d01d24cceb4ef : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17c2b525dbc44cc88e186f9749da09e5 : SUCCESS in 7m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed09ac314b8b4237a6cffebe3288a2b1 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":1},{"id":"65b1b7dc7f2cc7f43db9932298201d85b23566ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2021-08-06 07:47:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"876ae69e1b60cb4c340c53dd41ca31064ff1ee10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 09:31:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c4a6252272de40a785c996278126df1f : SUCCESS in 4m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4b6735c1590d4967a0051d058e692718 : SUCCESS in 1h 34m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b5f07bd032144c1985b8e3eb18c1a02a : SUCCESS in 1h 15m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e37206e19eaf4e908d2c75c6ba6b1f37 : SUCCESS in 4m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/448e33315bcc456bb604c5310ddd105c : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d63ee5c6069d4391a247c9902ded2fcf : SUCCESS in 5m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9503b61173b8487588df20da032ec8c1 : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/808bc238a8f14beaa5af6ded71d61b1f : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c978c274d1044bb5aa6227b8329b1af3 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":2},{"id":"f2caf91b29319865c217f3534252be49c0482375","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-06 09:47:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"833e0fbe113427a241d23ac501978930140400fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 09:47:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c569c1c73500b6523bdc9a602cc2a56d55ce25a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 11:13:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c152998e692449978c1c602b73540145 : SUCCESS in 4m 14s\n- tempest-full https://zuul.opendev.org/t/openstack/build/311d2b73e70a44b0b44963a9a051e22c : SUCCESS in 1h 26m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/976523aefbe84f36a0d9b3cbcdf37d50 : SUCCESS in 1h 18m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a06760758a2847b5bde3cf2a5b762428 : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/056475cd988f498091cb08e9f3acc0cc : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/554797ed65a344ce9f28e2e9cdd35945 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ca4563fd26f7460e8d8f3c1000825963 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07b94da1599e48e687b8a9655c2eedc2 : SUCCESS in 5m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8aa9e760213b4cd4b934f0407a7713b2 : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":2},{"id":"1c999f6a94a5b710c6b071a425d7868fb7e5d96a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 11:13:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7e5ba3581ec97efb98329be243a1cac80b7489b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 11:15:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/636578fe1eb24090a748d28020f504b2 : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2bbcfd9593d49f790373bf21aaa4a58 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4e16d837629c935e1a12f3bb55c77e1f64289ce0","revisions":{"c32a07bd44307192dbc7c3965d4d773817854d5b":{"kind":"REWORK","_number":1,"created":"2021-07-28 14:47:50.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/31/802731/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/31/802731/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/31/802731/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/31/802731/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/31/802731/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/31/802731/1"}}},"commit":{"parents":[{"commit":"c9f815d3b5ad066825e5bd8b6f10e2d44b5fb2f9","subject":"Dropping lower constraints testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/c9f815d3b5ad066825e5bd8b6f10e2d44b5fb2f9"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2021-02-19 06:35:08.000000000","tz":480},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-07-28 14:47:32.000000000","tz":60},"subject":"Remove lower-constraints remnants","message":"Remove lower-constraints remnants\n\nThese were missed in change : Icd090cf8a36fa560bf7b0c67d0e1c4923c221c65\n\nConflicts:\n  lower-constraints.txt\n\nNOTE(stephenfin): Conflicts are because the constraints and tox.ini are\nslightly different\n\nChange-Id: I07b577f4e948c03d1751145e5f898b90ac1c197a\n(cherry picked from commit ad720c3371ff26d484b5264f1332d92524719023)\n(cherry picked from commit 0646eb6595a79edb62801232ee2f6af0886d3013)\n(cherry picked from commit 1e40b7c4f3accd15dc48130b148141ae2740474c)\n(cherry picked from commit 14709ec56353f829dc1c21ef13d92f02e4c8e447)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/c32a07bd44307192dbc7c3965d4d773817854d5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/c32a07bd44307192dbc7c3965d4d773817854d5b"}]},"branch":"refs/heads/stable/train"},"4e16d837629c935e1a12f3bb55c77e1f64289ce0":{"kind":"REWORK","_number":2,"created":"2021-08-06 07:47:13.000000000","uploader":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"ref":"refs/changes/31/802731/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/31/802731/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/31/802731/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/31/802731/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/31/802731/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/31/802731/2"}}},"commit":{"parents":[{"commit":"c9f815d3b5ad066825e5bd8b6f10e2d44b5fb2f9","subject":"Dropping lower constraints testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/c9f815d3b5ad066825e5bd8b6f10e2d44b5fb2f9"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2021-02-19 06:35:08.000000000","tz":480},"committer":{"name":"Aija Jauntēva","email":"aija.jaunteva@dell.com","date":"2021-08-06 07:44:15.000000000","tz":-240},"subject":"Remove lower-constraints remnants","message":"Remove lower-constraints remnants\n\nThese were missed in change : Icd090cf8a36fa560bf7b0c67d0e1c4923c221c65\n\nConflicts:\n  lower-constraints.txt\n\nNOTE(stephenfin): Conflicts are because the constraints and tox.ini are\nslightly different\n\n(cherry picked from commit ad720c3371ff26d484b5264f1332d92524719023)\n(cherry picked from commit 0646eb6595a79edb62801232ee2f6af0886d3013)\n(cherry picked from commit 1e40b7c4f3accd15dc48130b148141ae2740474c)\n(cherry picked from commit d15545a7b3b745d8c23bcb5842b28ee0fecc6f66)\n\nNOTE(ajya): Added to fix mysql8 support for requirements-check job\n(cherry picked from commit 7af699e72802817d5a19fc0c4e016f6aeb9dce1c)\nChange-Id: I07b577f4e948c03d1751145e5f898b90ac1c197a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/4e16d837629c935e1a12f3bb55c77e1f64289ce0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/4e16d837629c935e1a12f3bb55c77e1f64289ce0"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
