)]}'
{"id":"openstack%2Foslo.db~804705","triplet_id":"openstack%2Foslo.db~master~I8629225eeb51d95264d8a3e4b719268bb1597f4f","project":"openstack/oslo.db","branch":"master","topic":"sqlalchemy-20","hashtags":[],"change_id":"I8629225eeb51d95264d8a3e4b719268bb1597f4f","subject":"requirements: Bump sqlalchemy lower constraint","status":"MERGED","created":"2021-08-16 11:26:26.000000000","updated":"2021-08-19 11:03:19.000000000","submitted":"2021-08-19 11:02:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"804705-sqlalchemy-20","meta_rev_id":"f761c59521ced6becccad87d8f506e8ab1f2344a","_number":804705,"virtual_id_number":804705,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-19 11:02:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":2,"date":"2021-08-16 17:47:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":2,"date":"2021-08-19 09:41:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":1,"date":"2021-08-19 09:41:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-16 12:40:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-16 17:47:12.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2021-08-19 09:41:41.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"fa8b409a1d17697c85df6bec54161c2ff0d1b25c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-16 11:26:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7bd55807ce16e446a5112b596d9b1f5d181e1a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 12:40:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/188a496861414726962213544aee78a1 : SUCCESS in 4m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/44337f1911d0435e8e44529dfdfd496f : SUCCESS in 1h 12m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/353acbd7d1144d10959f459fc8fb6d93 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15e4848a66af4c908da437810e3562f1 : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8ff907bb3994360a7af32cd6359deac : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/857863f5570e43bfa26bb6ee23489f72 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54c09f7d7adb403087e4d1554b9f59e6 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":1},{"id":"2a67d6b7016176f5c76670a7d4994201da2a0fcc","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2021-08-16 17:47:12.000000000","message":"Patch Set 1: Code-Review+2\n\nthis is fine, 1.4.0 is the target API wise.   it was loaded with regressions compared to 1.4.22 but i think that\u0027s not the point here.","accounts_in_message":[],"_revision_number":1},{"id":"a85a91b5a2762e08f66219080151fb86b2218154","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-19 09:41:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6c28c9a7949a3f3da6aea90e202c927e45bfc7ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 09:42:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a3f9557e3ea1ca40116fbebdb6eb71b0483114c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 11:02:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3b1ee5c53d2049da8674d86ce66203a4 : SUCCESS in 3m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8c0770c70c3443a097be95f7149b9d50 : SUCCESS in 1h 14m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/481446608f044df89c4c7853a65a9d5f : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25b07b88357644f7ba9e8be205a59a1e : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b58ad88b42448b78bbb2a3fc27ae108 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f21fb3d67c154700b11ed6c676c47a84 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":1},{"id":"1703d1f1f9b28f00adac01351e2d7fdc3455d8b2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 11:02:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f761c59521ced6becccad87d8f506e8ab1f2344a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 11:03:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d6914a855624b4984d53105abfbccde : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ad4729d69a6d4d74f27c24852f0e5674dd62c9a9","revisions":{"ad4729d69a6d4d74f27c24852f0e5674dd62c9a9":{"kind":"REWORK","_number":1,"created":"2021-08-16 11:26:26.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/05/804705/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/05/804705/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/05/804705/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/05/804705/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/05/804705/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/05/804705/1"}}},"commit":{"parents":[{"commit":"0816efca7b56032b10155bdf03daec3900b4ccd7","subject":"Update import of declarative_base()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/0816efca7b56032b10155bdf03daec3900b4ccd7"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-08-16 11:24:58.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-08-16 11:24:58.000000000","tz":60},"subject":"requirements: Bump sqlalchemy lower constraint","message":"requirements: Bump sqlalchemy lower constraint\n\nWe\u0027ve started using sqlalchemy features that were introduced in 1.4,\nwhich means we need to bump our lower constraint. Do this now.\n\nAt some point, we probably want to dump all lower-constraint information\nfrom these files since it\u0027s no longer maintainable without the\nequivalent job. That\u0027s a task left to later, however.\n\nChange-Id: I8629225eeb51d95264d8a3e4b719268bb1597f4f\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/ad4729d69a6d4d74f27c24852f0e5674dd62c9a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/ad4729d69a6d4d74f27c24852f0e5674dd62c9a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
