)]}'
{"id":"openstack%2Foslo.db~804775","triplet_id":"openstack%2Foslo.db~master~I7867cdcea115b13f2e45e0674bb9ef2ad138aae9","project":"openstack/oslo.db","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-07-20 15:55:26.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-07-20 15:55:26.000000000","reason":"Change was submitted"},"11816":{"account":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"last_update":"2022-07-20 14:15:58.000000000","reason":"\u003cGERRIT_ACCOUNT_11816\u003e replied on the change","reason_account":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}}},"hashtags":[],"change_id":"I7867cdcea115b13f2e45e0674bb9ef2ad138aae9","subject":"Remove the \u0027Session.autocommit\u0027 parameter","status":"MERGED","created":"2021-08-16 16:12:19.000000000","updated":"2022-09-04 07:57:22.000000000","submitted":"2022-07-20 15:55:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":24,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"804775-sqlalchemy-20","meta_rev_id":"2f31eea7cf26a751a7ce5a79f623572d6057f305","_number":804775,"virtual_id_number":804775,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-20 15:55:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2022-07-20 14:15:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2022-07-20 14:15:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-16 16:41:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-19 16:24:26.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"CC"},{"updated":"2022-02-10 12:16:51.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-07-20 14:15:58.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2022-09-04 07:57:22.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"4d3d4836b06d9bb9c344d0cd8cb17e4c38293c7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-16 16:12:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c614e7a01ca73a8ac3262538b69e897e1ab3df87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 16:41:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/23f3be996a2d4a679d0d04a70ffebab3 : FAILURE in 27m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f421685cc5b455386c66ccbb2f72f3c : FAILURE in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/faf75bf4459e44f8a6bc1fb94f3581fb : SUCCESS in 8m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e677b27f4c3f4c63a957264d0da7f991 : SUCCESS in 6m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8062124bc4b8401b85234387601eb3aa : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21c65b4ff8c04dcf88c2f7b7636774b9 : SUCCESS in 8m 35s\n\nWarning:\n  Comments left for invalid file oslo_db/tests/sqlalchemy/test_provision.py","accounts_in_message":[],"_revision_number":1},{"id":"71bf600ec023ba466619137e5398852d5ba1f456","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-16 16:56:06.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"9bd857c792a206ec284c758e08b150729bf281dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 17:34:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0851d253bb4b4d64b4771174ed07d0a5 : FAILURE in 32m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d58d7c2843384b08a71b14ed66d40b37 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7bb09aa43bb4c12bb15b46c7978c87c : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b21de53b43041e183abc1cf2d918af3 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8855b227d45a4fbd8d1ae4b9f0ce3279 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb5b05c81d1f46a394ecafae7e46a9ca : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":2},{"id":"dfa233c36f6f1bf8961a62cc60ee948dc1bb019a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-17 09:10:49.000000000","message":"Patch Set 2: Workflow-1\n\nThis is clearly breaking something somewhere, though I\u0027m not entirely sure what or where","accounts_in_message":[],"_revision_number":2},{"id":"29e21c9b3238c7942bfaa8bdd06099791e4309c9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-17 09:13:03.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2: Workflow-1\n\u003e \n\u003e This is clearly breaking something somewhere, though I\u0027m not entirely sure what or where\n\nAh, it\u0027s neutron [1]. Now, how does one fix this...?\n\n[1] https://zuul.opendev.org/t/openstack/build/0851d253bb4b4d64b4771174ed07d0a5/log/controller/logs/screen-q-svc.txt","accounts_in_message":[],"_revision_number":2},{"id":"a347d06420e4babe7dfcdcb61873367484be414a","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2021-08-19 16:24:26.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e \u003e Patch Set 2: Workflow-1\n\u003e \u003e \n\u003e \u003e This is clearly breaking something somewhere, though I\u0027m not entirely sure what or where\n\u003e \n\u003e Ah, it\u0027s neutron [1]. Now, how does one fix this...?\n\u003e \n\u003e [1] https://zuul.opendev.org/t/openstack/build/0851d253bb4b4d64b4771174ed07d0a5/log/controller/logs/screen-q-svc.txt\n\nneutron will have to change basically, there\u0027s no such thing as \"not in a transaction\" if they have a session in scope and im not sure what they have been doing there","accounts_in_message":[],"_revision_number":2},{"id":"c1ba703d3e1e82504e31243666d522355c8726ec","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2021-11-12 16:26:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"60d2ca136974037fa145341a1bdf1ac2fb3e63f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 17:07:32.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5857b530531244d6ad79bb8165baf20c : FAILURE in 31m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd948b80883e4773b321622cb3cbccd3 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a303a7f613dd4d959c304098d922fec5 : SUCCESS in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/644043584da64815ab4d2987bccf61f0 : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c12e7b98bcab4ff692a8908119ac6242 : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcdeff46eeb54c33a789924794acc1b5 : SUCCESS in 8m 14s","accounts_in_message":[],"_revision_number":2},{"id":"a132c86a57d9bc3f80ffed65030a52d5c0fef347","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2021-11-12 17:20:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5b52ce5de014c2415a2eab35620792d7b7adeded","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-02-08 14:46:49.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"21c70b26049ae470b6d593e2f7a9dd747e2b448b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 15:29:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/352e7f27cde64e0cbcf807c0092a5269 : FAILURE in 25m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c978bcadb9344d09d8559478e1104eb : SUCCESS in 7m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43b993f76107484ea072a78108d371d4 : SUCCESS in 6m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/833bb2cf5f1346d2aef1235744e9fd1f : SUCCESS in 8m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ac82e09f7064e6ea87ac1afc4331fbe : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":3},{"id":"5a6f5bc3443bfefc4b8bc89dd9a31c61be0b2867","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-02-08 16:32:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"685dc5e2d380d1fda958a4791de47ea352a07c28","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-02-09 15:25:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"be554d73ee433e8c170d5f1d32b7c459634751d6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-02-10 16:25:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8267b0b18fa8142c276bbc02981d3fc16bb3e3d5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-02-10 16:51:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3806f146c8ff0249d0664aaa5832ba10e117c74f","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2022-02-10 17:18:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1812e62d590550f9409f8040844ee505ec7a624e","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2022-02-10 17:20:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"499b6f2d7673e04f1686417dd973e9fb4f3f18ce","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-02-11 16:13:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c31c86744743b5eff941ae3d469a07cca6899ddc","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-02-11 18:22:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8a66ceaa47c8af9c934f8a4e412a1781a13a3f25","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2022-02-11 23:30:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"97879562d735e5d026c471cdcd25657a60d2ac62","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2022-02-12 17:58:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f48b84633d4adce50fb3e0aa8c294ba83ffb1493","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2022-02-12 18:16:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7013cfc5fc05769ca6b67d16b003046c4d15be76","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-02-23 17:39:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f81edc938ba84ca39136f003f3c8dfab5494363c","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2022-05-03 13:36:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2132b7c8e911270d8d59e101dbe737e1a71a2927","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 15:47:06.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1fc3e0b51b4849a0934299558c7b94ee : TIMED_OUT in 2h 04m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b789088d72b54ebc9ef709c2f563543d : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27f7aee7883046959d8132bfab1e8f28 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d4f1d9932cf4deb926780ba31ac061d : SUCCESS in 5m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a46d27ccf394ffab8fc3fc691be7063 : RETRY_LIMIT in 3m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/338bc11651134e99a03945052df64798 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":3},{"id":"a66c28181ee3c72e6299efbbc1b8bc208225d974","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-05-03 16:17:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cbdcea6192b19c1fe58cac8f258eefcb2e33660a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-05-03 16:17:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3705cd74661244023f39c72e0fff2e8d8eb33c25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 18:27:40.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/15d40adda302459aada648087dd768b1 : TIMED_OUT in 2h 04m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47db671e2b784ce7a059e34a1e056334 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8485c8633ad0432d9f98c86a59510a14 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e37780575064e97b8f00f4afacb5001 : SUCCESS in 5m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a75e68897a264f609577c54f38d6fe0c : RETRY_LIMIT in 2m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f40ff308871b412f96e84fda89d32b7b : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":3},{"id":"fbf0765c9d0b012662244c0ce2e36c9daec56948","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2022-05-03 18:29:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c0918df7566d6352579dd5abd8142947d46465d6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-05-04 14:35:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f9020d95635d5b65c23a77ed8e2baf5601918ce8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-05-06 08:15:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cb81ece8b59751b8993faf0f3128edd4745d5e53","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-14 10:25:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"00ad5ef6ca9b86eafed9598dfc88d8ee4a3296b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 12:26:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d2be449ccd34c17aac4d64872e74310\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3a3eb9dc72974e009303446eea8dacf4 : SUCCESS in 1h 53m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65b94a1079c74595abb0c91881f18154 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c5f7e4e159b4ef09230baa987997c1b : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/589d8dbbb06e47dd975f194500c7ed8e : SUCCESS in 5m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/719bb49033eb4145ab10b863396c9daf : FAILURE in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a742e063eb084c328440334ffef7af2b : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":3},{"id":"18b6a4906c8404e0917e5f4997128f0020bd7593","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2022-07-20 14:15:58.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"90e781ed917609d2cbf6a394b591b2a90c6f4c6b","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2022-07-20 14:16:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fd0c22b594e74b5244a1b41fc1861cec69902ac4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 14:16:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"96f15836305fc431cc921054e94c9f8f7d3c6845","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 15:55:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30f4b7c746044783a83bee9f851e4c59\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0f4fd6f0e47e4da291db7e0690ff7226 : SUCCESS in 1h 35m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d234f28f94c644a0800fe03481a84a0c : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3df38dc9db074101afa73a70d6870b19 : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ea5dd9fb65c49d4b973a8222804f09a : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/286924b5adfa49f1a7ed23e0081a3479 : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":3},{"id":"4e06397a58991e3a4eb0c3139bd06b324d83592d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 15:55:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"2f31eea7cf26a751a7ce5a79f623572d6057f305","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-09-04 07:57:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a530cbfcf2fadbc1bbd7a85086278100f15abf2e","revisions":{"fe08db797d5ecd2c7ed57d098e356a88a462f32e":{"kind":"REWORK","_number":1,"created":"2021-08-16 16:12:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/75/804775/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/75/804775/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/75/804775/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/75/804775/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/75/804775/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/75/804775/1"}}},"commit":{"parents":[{"commit":"519e62e04a46ede5c3de1963dbd2716f6693a3f1","subject":"utils: Remove troublesome utility methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/519e62e04a46ede5c3de1963dbd2716f6693a3f1"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-08-16 12:10:59.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-08-16 15:58:59.000000000","tz":60},"subject":"Remove the \u0027Session.autocommit\u0027 parameter","message":"Remove the \u0027Session.autocommit\u0027 parameter\n\nResolve the following RemovedIn20Warning warning:\n\n  The Session.autocommit parameter is deprecated and will be removed in\n  SQLAlchemy version 2.0.  The Session now features \"autobegin\" behavior\n  such that the Session.begin() method may be called if a transaction\n  has not yet been started yet. See the section session_explicit_begin\n  for background.\n\nChange-Id: I7867cdcea115b13f2e45e0674bb9ef2ad138aae9\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/fe08db797d5ecd2c7ed57d098e356a88a462f32e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/fe08db797d5ecd2c7ed57d098e356a88a462f32e"}]},"branch":"refs/heads/master"},"a62459d079bbcc7dee375ae158c061d74fdcfc37":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-08-16 16:56:06.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/75/804775/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/75/804775/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/75/804775/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/75/804775/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/75/804775/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/75/804775/2"}}},"commit":{"parents":[{"commit":"7013599c9e7c66e4d24ab01052153b994d593df9","subject":"utils: Remove troublesome utility methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7013599c9e7c66e4d24ab01052153b994d593df9"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-08-16 12:10:59.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-08-16 16:55:43.000000000","tz":60},"subject":"Remove the \u0027Session.autocommit\u0027 parameter","message":"Remove the \u0027Session.autocommit\u0027 parameter\n\nResolve the following RemovedIn20Warning warning:\n\n  The Session.autocommit parameter is deprecated and will be removed in\n  SQLAlchemy version 2.0.  The Session now features \"autobegin\" behavior\n  such that the Session.begin() method may be called if a transaction\n  has not yet been started yet. See the section session_explicit_begin\n  for background.\n\nChange-Id: I7867cdcea115b13f2e45e0674bb9ef2ad138aae9\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a62459d079bbcc7dee375ae158c061d74fdcfc37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a62459d079bbcc7dee375ae158c061d74fdcfc37"}]},"branch":"refs/heads/master"},"a530cbfcf2fadbc1bbd7a85086278100f15abf2e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-02-08 14:46:49.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/75/804775/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/75/804775/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/75/804775/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/75/804775/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/75/804775/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/75/804775/3"}}},"commit":{"parents":[{"commit":"43142792502afe5e5ce29a4cc3c291f673ef066e","subject":"Add missing \u0027connect\u0027 wrapper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/43142792502afe5e5ce29a4cc3c291f673ef066e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-08-16 12:10:59.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-02-08 14:46:42.000000000","tz":0},"subject":"Remove the \u0027Session.autocommit\u0027 parameter","message":"Remove the \u0027Session.autocommit\u0027 parameter\n\nResolve the following RemovedIn20Warning warning:\n\n  The Session.autocommit parameter is deprecated and will be removed in\n  SQLAlchemy version 2.0.  The Session now features \"autobegin\" behavior\n  such that the Session.begin() method may be called if a transaction\n  has not yet been started yet. See the section session_explicit_begin\n  for background.\n\nChange-Id: I7867cdcea115b13f2e45e0674bb9ef2ad138aae9\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a530cbfcf2fadbc1bbd7a85086278100f15abf2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a530cbfcf2fadbc1bbd7a85086278100f15abf2e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
