)]}'
{"id":"openstack%2Foslo.db~804777","triplet_id":"openstack%2Foslo.db~master~I7aed7789584bf0070f11c22b5eaa0e80c42dfc9c","project":"openstack/oslo.db","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-07-20 15:55:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7aed7789584bf0070f11c22b5eaa0e80c42dfc9c","subject":"trivial: Don\u0027t emit warnings for our own deprecations","status":"MERGED","created":"2021-08-16 16:12:19.000000000","updated":"2022-07-20 15:57:14.000000000","submitted":"2022-07-20 15:55:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"804777-sqlalchemy-20","meta_rev_id":"c83afabf01e664fbed143a409b900e26f336c7a8","_number":804777,"virtual_id_number":804777,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-20 15:55:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":2,"date":"2022-07-20 14:18:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":1,"date":"2022-07-20 14:18:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-16 16:38:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-20 14:18:51.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"e99f34293f235588c813653c2fffaa3a06933d5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-16 16:12:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d39ece1f8f3d94e86fed3c6d77386127e84986a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 16:38:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4cb784ca1e504e0b93e3f00a4369ad4f : FAILURE in 23m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4f60003e3354c9fb68d879c6d53a7c4 : FAILURE in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9b69144a4cc470fbcccbe554e0f6085 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d3e84cb19954940bd2e30212d42cff4 : SUCCESS in 7m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/940b703db35c4f17b5baf6c3709bde58 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1c4407ec7634af3822bc9315daa1cf1 : SUCCESS in 7m 33s\n\nWarning:\n  Comments left for invalid file oslo_db/tests/sqlalchemy/test_provision.py","accounts_in_message":[],"_revision_number":1},{"id":"23ec8d721d0cedcd5fca6be7e913c80bdf9e3b1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-16 16:56:06.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"1057e080641805681fd8f933b40806979e19bf02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 17:25:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9aab6c698c3545199f4c7f79d2c480cf : FAILURE in 26m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd0158c9eda54e5698bd7eb1bfac475c : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a6e3a72afb7437c877a97d454550637 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/699a25083e534d098f682a197862480c : SUCCESS in 7m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b95fead92e1e4b8c9121c1795d264ed2 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b68a517fa568493aa336836675b430b6 : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":2},{"id":"46b34c40ec9c104528945bec19244a2c96e560dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-02-08 14:46:49.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"2918d5d65f6ba9897b554d7f9487e3f2879df792","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 15:48:29.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/39a491ca2e044bc0ae3d9b5ca080c2e6 : FAILURE in 29m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24125a84e7de4629874ba98ab8813114 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e09a48e3ee094cf4a0bd04385a4ed8c3 : SUCCESS in 8m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e20349bb0af4a738832c72857b0a4d4 : SUCCESS in 7m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83f6899addaf4cbc8f38ae67c365b52d : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":3},{"id":"4264352db28beee163809c83fbf0949066c15ce0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-14 10:26:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"14cd33f4c8bf9b2e3d30ea8cdd5e5b670bbfd7c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 12:29:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e73223187dc34e088bb0600328d2c62d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cfa95488546140a28c4d40425c654fc5 : SUCCESS in 2h 02m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1578b094b914a28851027118ff7e457 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d118a5e978a5479691fd12e399a0f09f : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93bb850ca64e46f29b0cb85217ec8e9f : SUCCESS in 5m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/363b1570467d434ca163ce23cc1697be : FAILURE in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1936dfaf67ba48cdbd9b8c8885327d15 : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":3},{"id":"3bb60f6e544af06ca169eed4246050839233dbf0","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2022-07-20 14:18:51.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9c626429e9422df793db4fef7841bbca748e315e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 14:19:35.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0f90f1158f5d977c34e41da6547a99a873780e6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 15:55:29.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d645a9158fe4cb18b4280126b470b9b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c17e67b257204c10ac11937b43ca53f9 : SUCCESS in 1h 34m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18af4a23d53d4c4cac6f3e71fb5fb6f2 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0233df95d7eb46dcbed301dfefdbf3cb : SUCCESS in 7m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba5f2235f462466f9efeafe5ddd6c135 : SUCCESS in 7m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e79ccf0ace9c470b8d9f10ab6683124f : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":3},{"id":"f47d23a1a9a3188422a903210fd51876f4d7233a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 15:55:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"c83afabf01e664fbed143a409b900e26f336c7a8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 15:57:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2d29b3a24d44957825fb112dbc997a5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d92e1989e4594d2eb686845bf265bfd2 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4c451b7df55f3f77af7339ed38b49bdb5747e52e","revisions":{"440202bcf44018001ab2640c248d0818aedba269":{"kind":"REWORK","_number":1,"created":"2021-08-16 16:12:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/77/804777/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/77/804777/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/77/804777/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/77/804777/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/77/804777/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/77/804777/1"}}},"commit":{"parents":[{"commit":"9c90891577306dcd96d69eec6743292859cedfcb","subject":"tests: Enable SAWarning warnings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/9c90891577306dcd96d69eec6743292859cedfcb"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-08-16 15:19:23.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-08-16 16:11:38.000000000","tz":60},"subject":"trivial: Don\u0027t emit warnings for our own deprecations","message":"trivial: Don\u0027t emit warnings for our own deprecations\n\nWe\u0027ve deprecated a number of modules recently. We don\u0027t need to emit\nthese warnings when running unit tests. Silence things.\n\nChange-Id: I7aed7789584bf0070f11c22b5eaa0e80c42dfc9c\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/440202bcf44018001ab2640c248d0818aedba269"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/440202bcf44018001ab2640c248d0818aedba269"}]},"branch":"refs/heads/master"},"d0218915e02f76aded2cea6949c6f0070f3b7db4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-08-16 16:56:06.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/77/804777/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/77/804777/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/77/804777/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/77/804777/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/77/804777/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/77/804777/2"}}},"commit":{"parents":[{"commit":"763d81093664ab2b0fed51e958f793a04dc67c94","subject":"tests: Enable SAWarning warnings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/763d81093664ab2b0fed51e958f793a04dc67c94"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-08-16 15:19:23.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-08-16 16:55:56.000000000","tz":60},"subject":"trivial: Don\u0027t emit warnings for our own deprecations","message":"trivial: Don\u0027t emit warnings for our own deprecations\n\nWe\u0027ve deprecated a number of modules recently. We don\u0027t need to emit\nthese warnings when running unit tests. Silence things.\n\nChange-Id: I7aed7789584bf0070f11c22b5eaa0e80c42dfc9c\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/d0218915e02f76aded2cea6949c6f0070f3b7db4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/d0218915e02f76aded2cea6949c6f0070f3b7db4"}]},"branch":"refs/heads/master"},"4c451b7df55f3f77af7339ed38b49bdb5747e52e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-02-08 14:46:49.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/77/804777/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/77/804777/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/77/804777/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/77/804777/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/77/804777/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/77/804777/3"}}},"commit":{"parents":[{"commit":"2c7c0cab06adcfd5e510985dc1af302a47ae9899","subject":"tests: Enable SAWarning warnings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/2c7c0cab06adcfd5e510985dc1af302a47ae9899"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-08-16 15:19:23.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-02-08 14:46:42.000000000","tz":0},"subject":"trivial: Don\u0027t emit warnings for our own deprecations","message":"trivial: Don\u0027t emit warnings for our own deprecations\n\nWe\u0027ve deprecated a number of modules recently. We don\u0027t need to emit\nthese warnings when running unit tests. Silence things.\n\nChange-Id: I7aed7789584bf0070f11c22b5eaa0e80c42dfc9c\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/4c451b7df55f3f77af7339ed38b49bdb5747e52e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/4c451b7df55f3f77af7339ed38b49bdb5747e52e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
