)]}'
{"id":"openstack%2Foslo.db~804778","triplet_id":"openstack%2Foslo.db~master~Ib34d8ab411a67816db2e26b49bec75993b5bed56","project":"openstack/oslo.db","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-07-20 14:24:21.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"Ib34d8ab411a67816db2e26b49bec75993b5bed56","subject":"tox: Silence output","status":"MERGED","created":"2021-08-16 16:12:19.000000000","updated":"2022-07-20 16:02:38.000000000","submitted":"2022-07-20 16:01:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"804778-sqlalchemy-20","meta_rev_id":"82be4e7245c17e21a4eb7e955cffed22d14b424e","_number":804778,"virtual_id_number":804778,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-20 16:01:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":2,"date":"2022-07-20 14:20:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":1,"date":"2022-07-20 14:20:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-16 16:42:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-20 14:20:08.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"faf56fa5ca52966f290995d68bb78fc1f50374e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-16 16:12:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3211bc0946ae29489fc6dea69c3887a771a968ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 16:42:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5929ef1f668e4563aa89832772f30313 : FAILURE in 27m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a452861ca497459c8615b6e6fd3ebbb3 : FAILURE in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6145b6def19f47be90cae216fbe34578 : SUCCESS in 7m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae7b26279b6e43e6b3150d7b5d9e9922 : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6492314fe4ed4725b0e7fd615771f86d : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be8da8d29ccd42efa5807747a6707853 : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f1700cc85a44d378b2f58d62e6b0272 : SUCCESS in 6m 02s\n\nWarning:\n  Comments left for invalid file oslo_db/tests/sqlalchemy/test_provision.py","accounts_in_message":[],"_revision_number":1},{"id":"2a28d62900973b5f15ffa30b49d2bd8fe0b8ce52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-16 16:56:06.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"cef30d3497dbbbdb7b88430f9f12a542ded7d6e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 17:28:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4a74c3fd9e834ef4919c74c4912cffb7 : FAILURE in 27m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d61c841850c4c3eb2946910696677ec : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ac2dc71e5544340816d76dc411330e8 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce7571337cee4968b350f991cd51a80b : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb4ce51d806b437e977cc3246a2bb1c9 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92f45090ca154af2a18d75e415c0bf87 : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f8ab1109741402689c8135f1f663aa8 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":2},{"id":"40a87695cf15d24a9dddbf784eb9a717d189c98e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-02-08 14:46:49.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"db17a51b014a9222a873ef945d809fd00494b037","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 15:42:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9566b49eb1b648c897e1782f92e013e0 : FAILURE in 26m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd57dce3fd4a47e2b2d33819a0896f58 : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13d2d36fa4c24a6db853516f8a719e86 : SUCCESS in 7m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37a08049430743ca89ccc8ae7371b918 : SUCCESS in 7m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd7f459294424a4d8edcc5ec38c389c5 : SUCCESS in 8m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6958a2a8bfde4d35a6a015edb305e1a0 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":3},{"id":"5179eefcf670afd0689a06b5c499c11163ddaa26","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-14 10:26:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7fdd8eee90b8c2ba2d3635aa4e4675aeb1cc24a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 12:06:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2387016ac0948a996912f923dc391b3\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/24418ca313e94180ab0b13bdaa50612d : SUCCESS in 1h 38m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f09975c6ffb401ea98757afbaed70e6 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9f5f0801b874be592da90d6413a6214 : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/017af185e3ab4df0ab0b4c7ccc942d66 : SUCCESS in 5m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f6238defbc04997826ac705169f7baa : FAILURE in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f300b21bfd4448fbb4567409bc581f7 : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea85ccef13f44f879b97d7ee5b87b0b7 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":3},{"id":"c2f7b6c01466cd2f4c8b83c5f60ed0df4222263c","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2022-07-20 14:20:08.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2cbd7da43405412feeedac748f2123a1dd81d8f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 14:20:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6fee10fa72240dfade9fe1f51f52d647df507c30","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-20 14:24:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"17967912c82ae4b8f61d49b7ff45ce7f9f58da6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 16:01:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a219e608127e4f03b8c2f1b8881fd33f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/889d3ef9eacf4c6081c43c84c77d26d5 : SUCCESS in 1h 35m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6be76400871b4154a5070f2ba24c6a4f : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f24beb8fec846308e9d66160fe7eea2 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1e11914810f45908273dd2864b5123c : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b79681b23208463a97f5fdf7cc8524bf : SUCCESS in 4m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f03bdca5e564cf68ad2afcdc86b7751 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":3},{"id":"3ed0d6491c6aaef5b0e9c2e87ce086a29d812fe4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 16:01:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"82be4e7245c17e21a4eb7e955cffed22d14b424e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 16:02:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0ba597e7ab7495db2f78cff4bec3f19\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47d79706028e48bb9994c76325363365 : SUCCESS in 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1866ebda68f541a490e73d396238f320 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e7642fd2c8b8da1ecf18b914a4b773879861c6e8","revisions":{"3a4d6cb142c2b8941cd8a2a2eeb6446f8d3f2b50":{"kind":"REWORK","_number":1,"created":"2021-08-16 16:12:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/78/804778/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/78/804778/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/78/804778/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/78/804778/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/78/804778/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/78/804778/1"}}},"commit":{"parents":[{"commit":"440202bcf44018001ab2640c248d0818aedba269","subject":"trivial: Don\u0027t emit warnings for our own deprecations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/440202bcf44018001ab2640c248d0818aedba269"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-08-16 15:04:24.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-08-16 16:11:39.000000000","tz":60},"subject":"tox: Silence output","message":"tox: Silence output\n\nEnable the \u0027CaptureOutput\u0027 fixture provided by oslotest by default.\n\nChange-Id: Ib34d8ab411a67816db2e26b49bec75993b5bed56\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/3a4d6cb142c2b8941cd8a2a2eeb6446f8d3f2b50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/3a4d6cb142c2b8941cd8a2a2eeb6446f8d3f2b50"}]},"branch":"refs/heads/master"},"0a82902aa56d634eb7591fb20b01413d1f1c49e4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-08-16 16:56:06.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/78/804778/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/78/804778/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/78/804778/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/78/804778/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/78/804778/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/78/804778/2"}}},"commit":{"parents":[{"commit":"d0218915e02f76aded2cea6949c6f0070f3b7db4","subject":"trivial: Don\u0027t emit warnings for our own deprecations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/d0218915e02f76aded2cea6949c6f0070f3b7db4"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-08-16 15:04:24.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-08-16 16:55:57.000000000","tz":60},"subject":"tox: Silence output","message":"tox: Silence output\n\nEnable the \u0027CaptureOutput\u0027 fixture provided by oslotest by default.\n\nChange-Id: Ib34d8ab411a67816db2e26b49bec75993b5bed56\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/0a82902aa56d634eb7591fb20b01413d1f1c49e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/0a82902aa56d634eb7591fb20b01413d1f1c49e4"}]},"branch":"refs/heads/master"},"e7642fd2c8b8da1ecf18b914a4b773879861c6e8":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-02-08 14:46:49.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/78/804778/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/78/804778/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/78/804778/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/78/804778/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/78/804778/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/78/804778/3"}}},"commit":{"parents":[{"commit":"4c451b7df55f3f77af7339ed38b49bdb5747e52e","subject":"trivial: Don\u0027t emit warnings for our own deprecations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/4c451b7df55f3f77af7339ed38b49bdb5747e52e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-08-16 15:04:24.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-02-08 14:46:42.000000000","tz":0},"subject":"tox: Silence output","message":"tox: Silence output\n\nEnable the \u0027CaptureOutput\u0027 fixture provided by oslotest by default.\n\nChange-Id: Ib34d8ab411a67816db2e26b49bec75993b5bed56\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e7642fd2c8b8da1ecf18b914a4b773879861c6e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e7642fd2c8b8da1ecf18b914a4b773879861c6e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
