)]}'
{"id":"openstack%2Foslo.db~845110","triplet_id":"openstack%2Foslo.db~master~Ia4f3dcf410b2983ca0aca12f497739303d98112e","project":"openstack/oslo.db","branch":"master","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-04-17 09:45:19.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia4f3dcf410b2983ca0aca12f497739303d98112e","subject":"Removes option bindings deprecated since 2014","status":"ABANDONED","created":"2022-06-08 09:27:14.000000000","updated":"2023-04-17 09:45:19.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"d4ae6408df33d9f42a236e143ba668b1d5241327","_number":845110,"virtual_id_number":845110,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-06-08 11:08:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":-1,"date":"2022-06-21 16:18:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-08 11:08:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-21 16:18:54.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"952760b28e7ffd6e085bbef4b70d99389247f114","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-06-08 09:27:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"395c95c30f0e8a9acc0b9f2fec75a100476e3010","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-06-08 09:34:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0ca000cabff450cbf2df69167288e1ccdceb4883","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-06-08 09:37:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"657a05d521126c3b49b4eb14358f028f2a3d4777","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-06-08 09:38:22.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"cd8a4e1b9622814a1e929822316cf71217f31aa0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 11:08:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9c60b79ff021493e8c9188a67554bf34 : SUCCESS in 1h 29m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a512855b31b4e6aabbc142943aed8ca : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/234db5f3ad234c30a14c70acf3cf327f : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5008701d07154906a27d0ec59da1696a : SUCCESS in 5m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6956cb0e124440868caabc708ab7ca9c : FAILURE in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a653160e2ca540d294390ce83ef5e2d8 : SUCCESS in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de24f803619241778dbdb22909d77a4a : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":4},{"id":"bbacd1b0a7c2b50b64caa1643e328f0b5276edbc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-06-21 16:18:54.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d4ae6408df33d9f42a236e143ba668b1d5241327","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-04-17 09:45:19.000000000","message":"Abandoned\n\nThis has been in merge conflict for a long time. Please rebase and address the comments if this still applies. Thanks!","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"bd39da02ba09278c06aa524ee4608570f5cdb767","revisions":{"6eb348e734fcc1e8c63deaf4b0725567e7b5cd3c":{"kind":"REWORK","_number":1,"created":"2022-06-08 09:27:14.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/10/845110/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/10/845110/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/10/845110/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/10/845110/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/10/845110/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/10/845110/1"}}},"commit":{"parents":[{"commit":"3f889d5b6e4def3259e46a5ffe23c1350b4c10bf","subject":"Drop python3.6/3.7 support in testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/3f889d5b6e4def3259e46a5ffe23c1350b4c10bf"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-06-08 09:23:44.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-06-08 09:26:48.000000000","tz":120},"subject":"Remove option bindings deprecated since 2014","message":"Remove option bindings deprecated since 2014\n\nThose bindings are deprecated since 2014, I think\nwe can remove them safely.\n\nChange-Id: Ia4f3dcf410b2983ca0aca12f497739303d98112e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/6eb348e734fcc1e8c63deaf4b0725567e7b5cd3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/6eb348e734fcc1e8c63deaf4b0725567e7b5cd3c"}]},"branch":"refs/heads/master"},"d2bfa1ffa5ce8676523934f5d6579b4c4b03c0cf":{"kind":"REWORK","_number":2,"created":"2022-06-08 09:34:39.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/10/845110/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/10/845110/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/10/845110/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/10/845110/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/10/845110/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/10/845110/2"}}},"commit":{"parents":[{"commit":"3f889d5b6e4def3259e46a5ffe23c1350b4c10bf","subject":"Drop python3.6/3.7 support in testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/3f889d5b6e4def3259e46a5ffe23c1350b4c10bf"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-06-08 09:23:44.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-06-08 09:34:31.000000000","tz":120},"subject":"Remove option bindings deprecated since 2014","message":"Remove option bindings deprecated since 2014\n\nThose bindings are deprecated since 2014, I think\nwe can remove them safely.\n\nChange-Id: Ia4f3dcf410b2983ca0aca12f497739303d98112e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/d2bfa1ffa5ce8676523934f5d6579b4c4b03c0cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/d2bfa1ffa5ce8676523934f5d6579b4c4b03c0cf"}]},"branch":"refs/heads/master"},"0c0073a953be72e08f1aa69ca86617318717cb6b":{"kind":"REWORK","_number":3,"created":"2022-06-08 09:37:21.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/10/845110/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/10/845110/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/10/845110/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/10/845110/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/10/845110/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/10/845110/3"}}},"commit":{"parents":[{"commit":"3f889d5b6e4def3259e46a5ffe23c1350b4c10bf","subject":"Drop python3.6/3.7 support in testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/3f889d5b6e4def3259e46a5ffe23c1350b4c10bf"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-06-08 09:23:44.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-06-08 09:37:12.000000000","tz":120},"subject":"Remove option bindings deprecated since 2014","message":"Remove option bindings deprecated since 2014\n\nThose bindings are deprecated since 2014, I think\nwe can remove them safely.\n\nChange-Id: Ia4f3dcf410b2983ca0aca12f497739303d98112e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/0c0073a953be72e08f1aa69ca86617318717cb6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/0c0073a953be72e08f1aa69ca86617318717cb6b"}]},"branch":"refs/heads/master"},"bd39da02ba09278c06aa524ee4608570f5cdb767":{"kind":"REWORK","_number":4,"created":"2022-06-08 09:38:22.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/10/845110/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/10/845110/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/10/845110/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/10/845110/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/10/845110/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/10/845110/4"}}},"commit":{"parents":[{"commit":"3f889d5b6e4def3259e46a5ffe23c1350b4c10bf","subject":"Drop python3.6/3.7 support in testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/3f889d5b6e4def3259e46a5ffe23c1350b4c10bf"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-06-08 09:23:44.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-06-08 09:37:54.000000000","tz":120},"subject":"Removes option bindings deprecated since 2014","message":"Removes option bindings deprecated since 2014\n\nThose bindings are deprecated since 2014, I think\nwe can remove them safely.\n\nChange-Id: Ia4f3dcf410b2983ca0aca12f497739303d98112e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/bd39da02ba09278c06aa524ee4608570f5cdb767"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/bd39da02ba09278c06aa524ee4608570f5cdb767"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
