)]}'
{"id":"openstack%2Foslo.db~850536","triplet_id":"openstack%2Foslo.db~master~Id5ddf1d6f47b8a572001f58ad8b9b8a7dbe4e8ac","project":"openstack/oslo.db","branch":"master","topic":"deprecate-ndb-support","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-09-09 18:27:45.000000000","reason":"Change was submitted"},"11816":{"account":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"last_update":"2022-09-09 17:01:58.000000000","reason":"\u003cGERRIT_ACCOUNT_11816\u003e replied on the change","reason_account":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}}},"hashtags":[],"change_id":"Id5ddf1d6f47b8a572001f58ad8b9b8a7dbe4e8ac","subject":"Deprecate MySQL NDB Cluster Support","status":"MERGED","created":"2022-07-20 14:56:09.000000000","updated":"2022-09-09 18:29:36.000000000","submitted":"2022-09-09 18:27:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850536-deprecate-ndb-support","meta_rev_id":"bb45ae833c01de7fbe28e98248bbfb1f6d27ce25","_number":850536,"virtual_id_number":850536,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-09 18:27:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":2,"date":"2022-09-09 17:01:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":1,"date":"2022-09-09 17:01:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-20 15:06:36.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"CC"},{"updated":"2022-07-20 15:25:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-09 17:01:58.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"6fbcdcb872c9819f63d3640f9461fa40f633c290","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-20 14:56:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab0e281544535c66c1cf5db26bccda43afc88ea1","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2022-07-20 15:06:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2c0582116c51c935c822814723e99b3af05cbaa6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 15:25:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/002c6564de2246b892cab3280b85cf64\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e1d357ca23d4df9924c65adfce15b23 : FAILURE in 27m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb82325821fc4e22a766f916da3a37f0 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09b5e36279f349c0b84f4d527f81eacd : FAILURE in 8m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d84698020c341718192216f70d47189 : FAILURE in 7m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/718d669deb264fec806d559ad71f737c : FAILURE in 4m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d2a9d68a3664c0a9a77f094a228b7a5 : SUCCESS in 4m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5cb4d25b6cc54b1c96f49314d12b3f01 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":1},{"id":"e30c188c0d111ab350f524bd8d35ddb1dc1d42dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-27 10:30:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d528816e01ff8f4ff7d5c81103aeb219f6b2da2f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-27 10:30:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"24435791d5a65805108334736654866d1443a2f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 11:45:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f3657423c284ab5bd57497f0d54d313\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/886272851c434862ac1f5fdc920dd246 : SUCCESS in 1h 14m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf7a3b6e0afe489dad8dbe77fb174ec5 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a746496763a64a8882c619be1aebfce8 : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/399c7a6980f847908907a25f81aacdea : SUCCESS in 5m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e95a30badaa841cc92296894baf8ff64 : FAILURE in 4m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a792237642f34d40a4a5436e468b8976 : SUCCESS in 4m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2f0677924c34292a2a758366beffe38 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":2},{"id":"576ce1ef61192e329230429361903b232ff3722b","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2022-09-09 17:01:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4be58b0511344c913135739530e59c2e4748351f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 17:02:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a9c7d1dbff8808351a807ef173190d96874bdc5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 18:27:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba9689f312cf42dbb756297c1427162f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bae552bd1f6e477b9f9fd5b900ec9a46 : SUCCESS in 1h 21m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d365ad65ee9e484d99dcf08e9b21e549 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9faeb74d04a046a3921b7401975ed932 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eafe3d76a9974f149becb0c5976b4b89 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a53cb90b6c6473b994af258053a286e : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33e927a750e04f80a6af02ec8cb74164 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":2},{"id":"00b64e6c3ca3aaee182cb8168a0e266e52e9de75","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 18:27:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"bb45ae833c01de7fbe28e98248bbfb1f6d27ce25","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 18:29:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1610719b43a24e58a92ce2ffe00c6e2f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f42607121ca2423d83b12462c2bf4af6 : SUCCESS in 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a12ec66f19f4584ab2015d6faa50122 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f32890e41daca63b67074eb816b3d3d4458869b3","revisions":{"427d3a965df9a46f16ae16e3b908825a6094fc4c":{"kind":"REWORK","_number":1,"created":"2022-07-20 14:56:09.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/36/850536/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/36/850536/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/36/850536/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/36/850536/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/36/850536/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/36/850536/1"}}},"commit":{"parents":[{"commit":"8c9ef045d1e697c3bbca5379583f7bd93651f47f","subject":"trivial: Formatting changes for oslo_db.options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/8c9ef045d1e697c3bbca5379583f7bd93651f47f"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-07-14 10:33:19.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-07-20 14:55:51.000000000","tz":60},"subject":"Deprecate MySQL NDB Cluster Support","message":"Deprecate MySQL NDB Cluster Support\n\nNo one is using this engine and it\u0027s poorly understood. Deprecate it\nwith an eye on removing it sooner rather than later.\n\nChange-Id: Id5ddf1d6f47b8a572001f58ad8b9b8a7dbe4e8ac\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/427d3a965df9a46f16ae16e3b908825a6094fc4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/427d3a965df9a46f16ae16e3b908825a6094fc4c"}]},"branch":"refs/heads/master"},"f32890e41daca63b67074eb816b3d3d4458869b3":{"kind":"REWORK","_number":2,"created":"2022-07-27 10:30:01.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/36/850536/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/36/850536/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/36/850536/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/36/850536/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/36/850536/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/36/850536/2"}}},"commit":{"parents":[{"commit":"8c9ef045d1e697c3bbca5379583f7bd93651f47f","subject":"trivial: Formatting changes for oslo_db.options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/8c9ef045d1e697c3bbca5379583f7bd93651f47f"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-07-14 10:33:19.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-07-27 10:20:59.000000000","tz":60},"subject":"Deprecate MySQL NDB Cluster Support","message":"Deprecate MySQL NDB Cluster Support\n\nTraditionally, the MySQL support in oslo.db has assumed use of the\nInnoDB storage engine. However, this isn\u0027t the only storage engine\navailable and a few years ago an effort was made to add support for\nanother storage engine, MySQL Cluster (NDB). The oslo.db aspects of this\neffort were tracked via bug 1564110 [1] and from reading this bug and\nlooking at other patches related to this effort [2], it becomes obvious\nthat this was never seen through to the completion and the\nOpenStack-wide effort never took off [3]. As a result, much of what is\nhere is in-effect dead code now.\n\nGiven no one is using this engine, there\u0027s no reason to keep it around.\nDeprecate it with an eye on removing it sooner rather than later.\n\n[1] https://bugs.launchpad.net/oslo.db/+bug/1564110\n[2] https://review.opendev.org/q/owner:octave.orgeron%2540oracle.com\n[3] https://review.opendev.org/c/openstack/openstack-specs/+/429940\n\nChange-Id: Id5ddf1d6f47b8a572001f58ad8b9b8a7dbe4e8ac\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f32890e41daca63b67074eb816b3d3d4458869b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f32890e41daca63b67074eb816b3d3d4458869b3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
