)]}'
{"id":"openstack%2Foslo.db~850550","triplet_id":"openstack%2Foslo.db~stable%2Fwallaby~I27e20e59bce8c98b5a507c617364652aa6eb33f5","project":"openstack/oslo.db","branch":"stable/wallaby","topic":"modernize_update_match-stable/wallaby","attention_set":{},"removed_from_attention_set":{"1955":{"account":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"last_update":"2022-07-20 15:26:11.000000000","reason":"\u003cGERRIT_ACCOUNT_1955\u003e replied on the change","reason_account":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"}}},"hashtags":[],"change_id":"I27e20e59bce8c98b5a507c617364652aa6eb33f5","subject":"Don\u0027t use private API to get query criteria","status":"MERGED","created":"2022-07-20 15:22:32.000000000","updated":"2022-07-20 18:21:43.000000000","submitted":"2022-07-20 18:20:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850550-modernize_update_match-stable/wallaby","cherry_pick_of_change":758142,"cherry_pick_of_patch_set":2,"meta_rev_id":"809ed405bb6ffb871f5173810562cb1b564898cf","_number":850550,"virtual_id_number":850550,"owner":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-20 18:20:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":2,"date":"2022-07-20 15:25:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":1,"date":"2022-07-20 15:25:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-20 15:22:32.000000000","updated_by":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"CC"},{"updated":"2022-07-20 15:25:04.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2022-07-20 16:48:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0123154a1ece34638038ea35f6ed6305a81843c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"date":"2022-07-20 15:22:32.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"626a6700ec786e9fc7b4dc34c9e4647886878bdb","author":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"date":"2022-07-20 15:23:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3c48d416d7a369385cf7857c13de62e39a2c2bfe","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2022-07-20 15:25:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"809b0ce001a78097d846c3673ce894fe70c57cfa","author":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"date":"2022-07-20 15:26:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a8e49d45300e21415933baf48942d4afeab1db33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 16:48:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/285202286cfa431687dfdc8661310100\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4225c3fdfe0d401787f5cb9724d08546 : SUCCESS in 1h 23m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3aae2255d254160904e9a192db2ae59 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fcf6560cd896430592f30fff958fa131 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd0891a44cbe4e97a148168aa1428c2e : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc3f54fa59834194b2089e82615edd1d : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e5740f0221348a88fd6c3633658bc67 : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":1},{"id":"5121b483dd14d34fac902e3169d7310a52f3bdc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 16:49:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"70e69232670849f66b70b53ed7febfe1d9348b00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 18:20:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad09a1e14deb45d684c0a64bfb2c562e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c0d33ccb5a294d58a691c820c1514014 : SUCCESS in 1h 30m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce453db3e5fb4bed86e65920feefc7d6 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e17e6ec9e1454c7ca5cb4efb60b4ea6e : SUCCESS in 8m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40b9e87969f34986994df597783e4f91 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/137e526de55f401083ed2c19e9e9b753 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":1},{"id":"99012f10dda8d0b90302386508a09d4ecf02830f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 18:20:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"809ed405bb6ffb871f5173810562cb1b564898cf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 18:21:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7579aa7223c4117881e7fad662825bc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a23ef02b0e6e4a0bb6ecfc4b74f9c62d : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"308574e63520408fcbfae2c1d2f9a550b833d332","revisions":{"308574e63520408fcbfae2c1d2f9a550b833d332":{"kind":"REWORK","_number":1,"created":"2022-07-20 15:22:32.000000000","uploader":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"ref":"refs/changes/50/850550/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/50/850550/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/50/850550/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/50/850550/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/50/850550/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/50/850550/1"}}},"commit":{"parents":[{"commit":"3b452de5d2651d74732ea4548b03e37ffb8018f8","subject":"Merge \"Followup of \"Added handler for mysql 8.0.19 duplicate key error update\"\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/3b452de5d2651d74732ea4548b03e37ffb8018f8"}]}],"author":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2020-10-14 14:39:43.000000000","tz":-240},"committer":{"name":"Alan Pevec","email":"alan.pevec@redhat.com","date":"2022-07-20 15:22:32.000000000","tz":0},"subject":"Don\u0027t use private API to get query criteria","message":"Don\u0027t use private API to get query criteria\n\nAnother change for SQLAlchemy 1.4 is the mechanics\nof how the ``QueryContext`` object works.   Update the\nupdate_match extension such that it does not rely upon the\n``QueryContext`` object and uses public APIs instead.\n\nChange-Id: I27e20e59bce8c98b5a507c617364652aa6eb33f5\n(cherry picked from commit f2a94b6f1a8d1c2274ff81ae9968d22a22fd8503)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/308574e63520408fcbfae2c1d2f9a550b833d332"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/308574e63520408fcbfae2c1d2f9a550b833d332"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
