)]}'
{"id":"openstack%2Foslo.db~856838","triplet_id":"openstack%2Foslo.db~stable%2Fzed~I005af8c7afe3b5104068ca153aecbfe54e163ca3","project":"openstack/oslo.db","branch":"stable/zed","hashtags":[],"change_id":"I005af8c7afe3b5104068ca153aecbfe54e163ca3","subject":"types: Set \u0027cache_ok\u0027 (redux)","status":"MERGED","created":"2022-09-09 16:42:51.000000000","updated":"2022-12-20 19:24:23.000000000","submitted":"2022-12-20 19:23:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856838","meta_rev_id":"d54389f6a46c3be9f3c9b761732b6e0a2f56fe3f","_number":856838,"virtual_id_number":856838,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-20 19:23:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2022-12-20 17:48:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2022-12-20 17:48:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-09 16:53:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2ee7c2e1a907580c5b62ff868e4a9f4fc48db872","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-09 16:42:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"994e713aa1f44e95bf0441b87a54aa596607ad52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 16:53:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a3d4d45449e4817a95e740a33fa32b7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60ee327457db45b4a489e9503a932770 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ccbf68c92374b58acb4dbe68abf3b57 : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f9364f34a3784a87bcf0ee613e3d305c : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6226f767515410b9311c6df85b1c67d : FAILURE in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ada712265e44ad285bab6d2f6dff319 : SUCCESS in 7m 22s","accounts_in_message":[],"_revision_number":1},{"id":"cbc35b34c948d987f92a4dd6eac86e1e52bc4096","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-12-20 17:48:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2c6be63e017b82f2c453c5b7b8aa492068b2284f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 17:48:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9b7047d50d718c18732831c44ab92471a4265909","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 19:23:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfee396c01654684b46d9695bf51a089\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d1e4290cea0443ea641cfd320677843 : SUCCESS in 1h 33m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/593ba906465248f89c3209680fe3098f : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/baaa6f263653420fa1ae5c434bf26e0d : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15cc12cfb2434cdfaa92ef4a4e96e1e5 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a8101c9b4124af2b1fcbd073060e677 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":1},{"id":"4b273f87150457e7ef545a05d4e73a5435400c3a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 19:23:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d54389f6a46c3be9f3c9b761732b6e0a2f56fe3f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 19:24:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee9456458f9c4dc6b16f04f2e898f500\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/622b8a804e7047d39a687b91e146eeb6 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"717c824bf48dbad86611a1b34b585189e33fcc26","revisions":{"717c824bf48dbad86611a1b34b585189e33fcc26":{"kind":"REWORK","_number":1,"created":"2022-09-09 16:42:51.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/38/856838/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/38/856838/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/38/856838/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/38/856838/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/38/856838/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/38/856838/1"}}},"commit":{"parents":[{"commit":"95c2fc600283ca1466b40ca8d0bf8d45f5d06b79","subject":"Update TOX_CONSTRAINTS_FILE for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/95c2fc600283ca1466b40ca8d0bf8d45f5d06b79"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-09-08 09:45:05.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-09-09 16:42:41.000000000","tz":60},"subject":"types: Set \u0027cache_ok\u0027 (redux)","message":"types: Set \u0027cache_ok\u0027 (redux)\n\nIn change Idf6fd858fad9521c7c5ba82c31b6d3077756abd9, we defined the\n\u0027cache_ok\u0027 key on a number of custom types. However, we relied on this\nattribute being inheritable which is not the case due to how that\nattribute is checked for [1]. As a result, we must set this attribute on\nevery one of our custom types.\n\n[1] https://github.com/sqlalchemy/sqlalchemy/blob/rel_1_4_41/lib/sqlalchemy/sql/type_api.py#L984\n\nChange-Id: I005af8c7afe3b5104068ca153aecbfe54e163ca3\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n(cherry picked from commit 10e8e77a92455fb638dcc7e73c044cd22acd84ad)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/717c824bf48dbad86611a1b34b585189e33fcc26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/717c824bf48dbad86611a1b34b585189e33fcc26"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
