)]}'
{"id":"openstack%2Foslo.db~862429","triplet_id":"openstack%2Foslo.db~master~I34fcffdf260d6ae62b9733d2fe69906e3a6205bc","project":"openstack/oslo.db","branch":"master","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2022-10-25 14:46:02.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I34fcffdf260d6ae62b9733d2fe69906e3a6205bc","subject":"DNM: Test CockroachDB","status":"ABANDONED","created":"2022-10-23 14:41:24.000000000","updated":"2022-10-25 14:46:02.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"51095e75a39bc36b5ab3a334457801fd8baf9685","_number":862429,"virtual_id_number":862429,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-10-24 08:59:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-23 17:22:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0b3509cb413e6ab4a57545f641c0e4e3a7cf4084","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-10-23 14:41:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e02704d215bed57689f6b1b094d86ed15e251476","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-10-23 15:59:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"930b3f30f9770cfb09b5ab14ceff4e6b72cf0275","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-23 17:22:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1bad909fbbd435f8af3c417bd6f0145\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/60fd03f3af084516b432571ca6ef8edf : SUCCESS in 2m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3eade6bc4656499dbd96c27ddc0b4d55 : SUCCESS in 1h 13m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8cc71569c344b5ca1952e2b2024830b : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/583dcec935ce499787dc612b5f707de7 : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3c0ec1d737354d43be9d5eb7b38231a3 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ade7b25545424437a303adf8286d0a80 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":2},{"id":"26ae23e3e3494222ddb8a70256ba795b0ee2367b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-10-23 19:05:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4f51df705ebcc295018ce97d159cefad294f8d65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-10-23 19:46:56.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e81350879bce62b3ae08ec380dc6c3f694b995e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-10-23 20:17:22.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"44be3808e14c7b23f5d4c4c051167d0a3c287352","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-10-23 20:28:28.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"fc0ecbec6ae6e877e8c06242600c0c2798114c68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-23 21:43:16.000000000","message":"Patch Set 6: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d19503c6121644b9808acfdad50c7ac3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a6c741b7451341e2aec21a58b52839c8 : SUCCESS in 2m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/403b5257d5214927ac295d5cd394108d : SUCCESS in 1h 12m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7c02fb7eab64d56b5344158e48eaa3c : FAILURE in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2047a126b3104171b18aab4b1780c2d5 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d59c3e1492294a6f97ae39963c1e7931 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/424ce1b89ce34a5086916e18d6fac5be : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":6},{"id":"815d73662a8e6ec677707761447b5f1044b42273","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-10-24 07:17:54.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"d91f3d6b776b9770fad8a46b6214e1d7358c33d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 08:59:27.000000000","message":"Patch Set 7: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d745a46d0d24a9cae00d62c83ce1b59\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5e2282b4fc5e46f99c23688c437a26d8 : SUCCESS in 1m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/56783da1d6cc435d9ed29a1d46db5c1b : SUCCESS in 1h 36m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78065c53d42a4680960e5aac76ed2ccd : FAILURE in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09047cc0808942629b50b90113043d6a : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d0b8c2e9ebe64d0289e3cfe9f46f601e : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf1cd72335324ed4b5e6ba56d56061eb : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":7},{"id":"51095e75a39bc36b5ab3a334457801fd8baf9685","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-10-25 14:46:02.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"496526c4729e9db8c86d7c2a8d7075fd73df5156","revisions":{"12f82d5b59ddb2bea617064028fb271eecef277b":{"kind":"REWORK","_number":1,"created":"2022-10-23 14:41:24.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/29/862429/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/29/862429/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/29/862429/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/29/862429/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/29/862429/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/29/862429/1"}}},"commit":{"parents":[{"commit":"a191d2e629f55c999b37a7a795ad6fbee2807b98","subject":"Merge \"tests: Define a primary key\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a191d2e629f55c999b37a7a795ad6fbee2807b98"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2022-10-23 14:41:10.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2022-10-23 14:41:22.000000000","tz":120},"subject":"DNM: Test CockroachDB","message":"DNM: Test CockroachDB\n\nChange-Id: I34fcffdf260d6ae62b9733d2fe69906e3a6205bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/12f82d5b59ddb2bea617064028fb271eecef277b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/12f82d5b59ddb2bea617064028fb271eecef277b"}]},"branch":"refs/heads/master"},"bb2def40dbf18bd14da016db093fbb9e135b83a2":{"kind":"REWORK","_number":2,"created":"2022-10-23 15:59:43.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/29/862429/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/29/862429/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/29/862429/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/29/862429/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/29/862429/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/29/862429/2"}}},"commit":{"parents":[{"commit":"a191d2e629f55c999b37a7a795ad6fbee2807b98","subject":"Merge \"tests: Define a primary key\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a191d2e629f55c999b37a7a795ad6fbee2807b98"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2022-10-23 14:41:10.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-10-23 15:59:42.000000000","tz":0},"subject":"DNM: Test CockroachDB","message":"DNM: Test CockroachDB\n\nChange-Id: I34fcffdf260d6ae62b9733d2fe69906e3a6205bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/bb2def40dbf18bd14da016db093fbb9e135b83a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/bb2def40dbf18bd14da016db093fbb9e135b83a2"}]},"branch":"refs/heads/master"},"6ce9895c8f1bb1d7442349cb3b1f7040c8aa186b":{"kind":"REWORK","_number":3,"created":"2022-10-23 19:05:38.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/29/862429/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/29/862429/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/29/862429/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/29/862429/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/29/862429/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/29/862429/3"}}},"commit":{"parents":[{"commit":"a191d2e629f55c999b37a7a795ad6fbee2807b98","subject":"Merge \"tests: Define a primary key\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a191d2e629f55c999b37a7a795ad6fbee2807b98"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2022-10-23 14:41:10.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-10-23 19:05:32.000000000","tz":0},"subject":"DNM: Test CockroachDB","message":"DNM: Test CockroachDB\n\nChange-Id: I34fcffdf260d6ae62b9733d2fe69906e3a6205bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/6ce9895c8f1bb1d7442349cb3b1f7040c8aa186b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/6ce9895c8f1bb1d7442349cb3b1f7040c8aa186b"}]},"branch":"refs/heads/master"},"e258f291c99265664cdfb80f3f48532613bafccb":{"kind":"REWORK","_number":4,"created":"2022-10-23 19:46:56.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/29/862429/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/29/862429/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/29/862429/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/29/862429/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/29/862429/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/29/862429/4"}}},"commit":{"parents":[{"commit":"a191d2e629f55c999b37a7a795ad6fbee2807b98","subject":"Merge \"tests: Define a primary key\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a191d2e629f55c999b37a7a795ad6fbee2807b98"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2022-10-23 14:41:10.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-10-23 19:46:50.000000000","tz":0},"subject":"DNM: Test CockroachDB","message":"DNM: Test CockroachDB\n\nChange-Id: I34fcffdf260d6ae62b9733d2fe69906e3a6205bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e258f291c99265664cdfb80f3f48532613bafccb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e258f291c99265664cdfb80f3f48532613bafccb"}]},"branch":"refs/heads/master"},"7b1ed1abd6148b2ac6cd3152fe7edb1bc6bf81f0":{"kind":"REWORK","_number":5,"created":"2022-10-23 20:17:22.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/29/862429/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/29/862429/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/29/862429/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/29/862429/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/29/862429/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/29/862429/5"}}},"commit":{"parents":[{"commit":"a191d2e629f55c999b37a7a795ad6fbee2807b98","subject":"Merge \"tests: Define a primary key\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a191d2e629f55c999b37a7a795ad6fbee2807b98"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2022-10-23 14:41:10.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-10-23 20:17:16.000000000","tz":0},"subject":"DNM: Test CockroachDB","message":"DNM: Test CockroachDB\n\nChange-Id: I34fcffdf260d6ae62b9733d2fe69906e3a6205bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7b1ed1abd6148b2ac6cd3152fe7edb1bc6bf81f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7b1ed1abd6148b2ac6cd3152fe7edb1bc6bf81f0"}]},"branch":"refs/heads/master"},"a44c9e68ece3af9c9e91fd9a58e90ac5b5c53df0":{"kind":"REWORK","_number":6,"created":"2022-10-23 20:28:28.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/29/862429/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/29/862429/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/29/862429/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/29/862429/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/29/862429/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/29/862429/6"}}},"commit":{"parents":[{"commit":"a191d2e629f55c999b37a7a795ad6fbee2807b98","subject":"Merge \"tests: Define a primary key\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a191d2e629f55c999b37a7a795ad6fbee2807b98"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2022-10-23 14:41:10.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-10-23 20:28:23.000000000","tz":0},"subject":"DNM: Test CockroachDB","message":"DNM: Test CockroachDB\n\nChange-Id: I34fcffdf260d6ae62b9733d2fe69906e3a6205bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a44c9e68ece3af9c9e91fd9a58e90ac5b5c53df0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a44c9e68ece3af9c9e91fd9a58e90ac5b5c53df0"}]},"branch":"refs/heads/master"},"496526c4729e9db8c86d7c2a8d7075fd73df5156":{"kind":"REWORK","_number":7,"created":"2022-10-24 07:17:54.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/29/862429/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/29/862429/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/29/862429/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/29/862429/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/29/862429/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/29/862429/7"}}},"commit":{"parents":[{"commit":"a191d2e629f55c999b37a7a795ad6fbee2807b98","subject":"Merge \"tests: Define a primary key\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a191d2e629f55c999b37a7a795ad6fbee2807b98"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2022-10-23 14:41:10.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-10-24 07:17:49.000000000","tz":0},"subject":"DNM: Test CockroachDB","message":"DNM: Test CockroachDB\n\nChange-Id: I34fcffdf260d6ae62b9733d2fe69906e3a6205bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/496526c4729e9db8c86d7c2a8d7075fd73df5156"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/496526c4729e9db8c86d7c2a8d7075fd73df5156"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
