)]}'
{"id":"openstack%2Foslo.db~865877","triplet_id":"openstack%2Foslo.db~master~Iee7afcac8ba952a2d67a9ad9dd0e4eae3f42518e","project":"openstack/oslo.db","branch":"master","topic":"add_wsrep_sync_wait","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-12-08 11:53:35.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"11816":{"account":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"last_update":"2022-12-08 13:40:13.000000000","reason":"Change was submitted"},"20778":{"account":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"last_update":"2022-12-08 13:40:13.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-12-08 13:40:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iee7afcac8ba952a2d67a9ad9dd0e4eae3f42518e","subject":"Add option for wsrep_sync_wait","status":"MERGED","created":"2022-11-28 18:28:54.000000000","updated":"2022-12-08 13:41:16.000000000","submitted":"2022-12-08 13:40:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"865877-add_wsrep_sync_wait","meta_rev_id":"3d421d12ab9836579bfacb7c36d8dcdf1ed3c08f","_number":865877,"virtual_id_number":865877,"owner":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-08 13:40:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2022-12-08 11:53:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2022-12-05 18:58:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2022-12-08 11:53:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"reviewers":{"REVIEWER":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-28 19:42:15.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-11-28 19:42:15.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2022-11-28 20:14:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-28 21:28:55.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"state":"REVIEWER"}],"messages":[{"id":"ab9e7aab3c1cba8ad6cffdb3968930926ce3d596","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2022-11-28 18:28:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cedc13b2155cccab2fa074dfdd77dd1e60390cfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2022-11-28 18:30:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"15bf1acb5d8742383ecc1fd1022b13894474bdf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 20:14:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e4e1a5386e94c0baeb3d0539a773847\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c4a7d94f7ffe4a988431935cdb766179 : SUCCESS in 1h 39m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf1b8edc80d44453abe50e4a1d1a9055 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f457716dd8b1424cacff08afb366fa60 : FAILURE in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff0609f050bf489982655d6f10ec6005 : FAILURE in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9fd2e4dcf7e44299585dca2fd47cf6f : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2818f5a59ef74bbe80da23f97084a48b : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":2},{"id":"6cc6a38886279ca6eaf53e4269db52c86b40160a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2022-11-28 21:28:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4475895052bc732271b3cc7fc2d30fe4fa7161a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 23:26:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7cae579ca2b4145b42d263fefae1db8\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0a9ed38472ff4ca89cbe78da0af35f0a : SUCCESS in 1h 52m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec51632c3ee04fd382550d6fa42c7c99 : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3f92009a42a4e39aed904ae7a7339bc : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d7ec0903a9bb486491165e5bed370e81 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dfd6cd62ba44268a459f6c56cdc2b81 : SUCCESS in 3m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/400f0812094444a0959e01b6206161b6 : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":3},{"id":"33026cdcf5b892cb7b964fb53a66e4ad33abb105","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-11-30 11:57:55.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"eec408fe5e4f1d30d11ac6e3f132836e5680bcbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2022-12-05 16:49:48.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"624f885ff4920f3e294e293c92b3eeb78dbd9c5a","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2022-12-05 16:49:53.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"8561a194772e63bdcaa3b54da6ffac25aa704248","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-05 18:19:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1438a05cb9e497ca9cc2004406af5a9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b60aff68404e48028b065d27e36636ed : SUCCESS in 1h 28m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d28c1df531bc4d9ba949f6ca31c8a001 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41dffb2b187c496cbe28b03ed3fbc138 : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c269342cbf4341db9df3a16a3cc58bd7 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85c92f9d1fed4dc188afefc937a993ee : SUCCESS in 3m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5325fb4a3934849ae57a394cd0265a3 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":4},{"id":"aa40d365e273db4bab02fdd082ad725dddcc66b4","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2022-12-05 18:58:01.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"dfe2841cbe8c0b122bba1962143e74ffe6ff94fe","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-12-08 11:53:35.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"3c69c29c7c77a2f9fd782fb656ce90475d55c0fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 11:54:03.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8b8ce2fb7d290f7052316a0c760a8bb93faf0122","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 13:40:13.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd899e33b9d14e05a9d3e2d6bd0503f3\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/932af22a54a84450ae016aa8ffa75b73 : SUCCESS in 1h 45m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c2d20509b554076a0636b8832f2163d : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e57861d454a04f72945a2f616830bf8c : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5841e364fd9047f29f7e10b29cffdc82 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0d3058d7e344806983032b697d6bfee : SUCCESS in 4m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7b3263aa4c74fcfbbb819a53cf32004 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":4},{"id":"a5008a5baf3568718aa05df2c97241f2daaaf136","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 13:40:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"3d421d12ab9836579bfacb7c36d8dcdf1ed3c08f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 13:41:16.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eba6a3527fce4a4699aa664808fdd301\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdaab14a3014426cb9362f77b09ba237 : SUCCESS in 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e05424494db545e2ac034dfa29ebb1b1 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"009d23df45969036c70e4cf59eb4019aaace9a55","revisions":{"f65210c021741ba9895ec4b8b9113d86f6fa54d1":{"kind":"REWORK","_number":1,"created":"2022-11-28 18:28:54.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/77/865877/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/77/865877/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/77/865877/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/77/865877/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/77/865877/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/77/865877/1"}}},"commit":{"parents":[{"commit":"a191d2e629f55c999b37a7a795ad6fbee2807b98","subject":"Merge \"tests: Define a primary key\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a191d2e629f55c999b37a7a795ad6fbee2807b98"}]}],"author":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2022-11-28 17:20:46.000000000","tz":-300},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2022-11-28 18:26:02.000000000","tz":-300},"subject":"Add option for wsrep_sync_wait","message":"Add option for wsrep_sync_wait\n\nWhen using Galera, the wsrep_sync_wait option [1]\ncan change the behavior of a variety of Galera DQL/DML statements\nsuch that a particular operation will pause until outstanding\nwrite-sets are fully persisted to the local node.  The setting\nsupersedes the previous boolean parameter wsrep_causal_reads\nwhich only affected SELECT, with an updated approach that allows\nfor fine-grained control of so-called \"causality checks\"\non individual statement types.  The legacy-compatible setting\nof \u00271\u0027 indicates that READ/SELECT/BEGIN operations should\nproceed only after any pending writesets are fully available.\n\nThe use case for this setting is for an application that\nis running operations on multiple Galera nodes simultaenously.\nAn application that commits data on one node, and then immediately\nuses a different connection (on a potentially different node)\nto SELECT that data, may fail to see those changes if\n\"causality checks\" for SELECT are not enabled.  While\na COMMIT operation in Galera will block locally until all other\nnodes approve of the writeset, the operation does not block\nfor the subsequent period of time when other nodes are actually\npersisting that writeset.  Setting up \"causal reads\"\nin this case indicates that a SELECT operation will wait until\nany writesets in progress are available, thus maintaining\nserialization between the COMMIT and subsequent SELECT.\n\nAs the name implies, wsrep_sync_wait adds...waiting!  to the\noperation, and thus directly impacts performance by adding\nlatency to SELECT operations or to the operations that have\nbeen selected for causality checks, to the degree that\nconcurrent writesets are expected to be present.\n\nSince it\u0027s not expected that most if any Openstack applications\nactually need this setting in order to be effective with\nGalera multi-master operation, and as the setting is available\nwithin client session scope and also impacts performance,\nmaking it available on a per-application basis means that\nspecific applications which may see issues under load can\nchoose to enable this setting, much in the way any other\n\"transaction isolation\" settings might be made, without having\nto add a cluster-wide performance penalty by setting it at the\nGalera server level.\n\n[1] https://mariadb.com/docs/ent/ref/mdb/system-variables/wsrep_sync_wait/\n\nChange-Id: Iee7afcac8ba952a2d67a9ad9dd0e4eae3f42518e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f65210c021741ba9895ec4b8b9113d86f6fa54d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f65210c021741ba9895ec4b8b9113d86f6fa54d1"}]},"branch":"refs/heads/master"},"0eda37c52e7f354e3776d070a782a0714fa9dd8b":{"kind":"REWORK","_number":2,"created":"2022-11-28 18:30:03.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/77/865877/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/77/865877/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/77/865877/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/77/865877/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/77/865877/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/77/865877/2"}}},"commit":{"parents":[{"commit":"a191d2e629f55c999b37a7a795ad6fbee2807b98","subject":"Merge \"tests: Define a primary key\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a191d2e629f55c999b37a7a795ad6fbee2807b98"}]}],"author":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2022-11-28 17:20:46.000000000","tz":-300},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2022-11-28 18:29:57.000000000","tz":-300},"subject":"Add option for wsrep_sync_wait","message":"Add option for wsrep_sync_wait\n\nWhen using Galera, the wsrep_sync_wait option [1]\ncan change the behavior of a variety of Galera DQL/DML statements\nsuch that a particular operation will pause until outstanding\nwrite-sets are fully persisted to the local node.  The setting\nsupersedes the previous boolean parameter wsrep_causal_reads\nwhich only affected SELECT, with an updated approach that allows\nfor fine-grained control of so-called \"causality checks\"\non individual statement types.  The legacy-compatible setting\nof \u00271\u0027 indicates that READ/SELECT/BEGIN operations should\nproceed only after any pending writesets are fully available.\n\nThe use case for this setting is for an application that\nis running operations on multiple Galera nodes simultaenously.\nAn application that commits data on one node, and then immediately\nuses a different connection (on a potentially different node)\nto SELECT that data, may fail to see those changes if\n\"causality checks\" for SELECT are not enabled.  While\na COMMIT operation in Galera will block locally until all other\nnodes approve of the writeset, the operation does not block\nfor the subsequent period of time when other nodes are actually\npersisting that writeset.  Setting up \"causal reads\"\nin this case indicates that a SELECT operation will wait until\nany writesets in progress are available, thus maintaining\nserialization between the COMMIT and subsequent SELECT.\n\nAs the name implies, wsrep_sync_wait adds...waiting!  to the\noperation, and thus directly impacts performance by adding\nlatency to SELECT operations or to the operations that have\nbeen selected for causality checks, to the degree that\nconcurrent writesets are expected to be present.\n\nSince it\u0027s not expected that most if any Openstack applications\nactually need this setting in order to be effective with\nGalera multi-master operation, and as the setting is available\nwithin client session scope and also impacts performance,\nmaking it available on a per-application basis means that\nspecific applications which may see issues under load can\nchoose to enable this setting, much in the way any other\n\"transaction isolation\" settings might be made, without having\nto add a cluster-wide performance penalty by setting it at the\nGalera server level.\n\n[1] https://mariadb.com/docs/ent/ref/mdb/system-variables/wsrep_sync_wait/\n\nChange-Id: Iee7afcac8ba952a2d67a9ad9dd0e4eae3f42518e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/0eda37c52e7f354e3776d070a782a0714fa9dd8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/0eda37c52e7f354e3776d070a782a0714fa9dd8b"}]},"branch":"refs/heads/master"},"706b364dc7a26c1f7c29c1fe931c80c85813a263":{"kind":"REWORK","_number":3,"created":"2022-11-28 21:28:15.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/77/865877/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/77/865877/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/77/865877/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/77/865877/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/77/865877/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/77/865877/3"}}},"commit":{"parents":[{"commit":"a191d2e629f55c999b37a7a795ad6fbee2807b98","subject":"Merge \"tests: Define a primary key\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a191d2e629f55c999b37a7a795ad6fbee2807b98"}]}],"author":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2022-11-28 17:20:46.000000000","tz":-300},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2022-11-28 21:27:26.000000000","tz":-300},"subject":"Add option for wsrep_sync_wait","message":"Add option for wsrep_sync_wait\n\nWhen using Galera, the wsrep_sync_wait option [1]\ncan change the behavior of a variety of Galera DQL/DML statements\nsuch that a particular operation will pause until outstanding\nwrite-sets are fully persisted to the local node.  The setting\nsupersedes the previous boolean parameter wsrep_causal_reads\nwhich only affected SELECT, with an updated approach that allows\nfor fine-grained control of so-called \"causality checks\"\non individual statement types.  The legacy-compatible setting\nof \u00271\u0027 indicates that READ/SELECT/BEGIN operations should\nproceed only after any pending writesets are fully available.\n\nThe use case for this setting is for an application that\nis running operations on multiple Galera nodes simultaenously.\nAn application that commits data on one node, and then immediately\nuses a different connection (on a potentially different node)\nto SELECT that data, may fail to see those changes if\n\"causality checks\" for SELECT are not enabled.  While\na COMMIT operation in Galera will block locally until all other\nnodes approve of the writeset, the operation does not block\nfor the subsequent period of time when other nodes are actually\npersisting that writeset.  Setting up \"causal reads\"\nin this case indicates that a SELECT operation will wait until\nany writesets in progress are available, thus maintaining\nserialization between the COMMIT and subsequent SELECT.\n\nAs the name implies, wsrep_sync_wait adds...waiting!  to the\noperation, and thus directly impacts performance by adding\nlatency to SELECT operations or to the operations that have\nbeen selected for causality checks, to the degree that\nconcurrent writesets are expected to be present.\n\nSince it\u0027s not expected that most if any Openstack applications\nactually need this setting in order to be effective with\nGalera multi-master operation, and as the setting is available\nwithin client session scope and also impacts performance,\nmaking it available on a per-application basis means that\nspecific applications which may see issues under load can\nchoose to enable this setting, much in the way any other\n\"transaction isolation\" settings might be made, without having\nto add a cluster-wide performance penalty by setting it at the\nGalera server level.\n\n[1] https://mariadb.com/docs/ent/ref/mdb/system-variables/wsrep_sync_wait/\n\nChange-Id: Iee7afcac8ba952a2d67a9ad9dd0e4eae3f42518e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/706b364dc7a26c1f7c29c1fe931c80c85813a263"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/706b364dc7a26c1f7c29c1fe931c80c85813a263"}]},"branch":"refs/heads/master"},"009d23df45969036c70e4cf59eb4019aaace9a55":{"kind":"REWORK","_number":4,"created":"2022-12-05 16:49:48.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/77/865877/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/77/865877/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/77/865877/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/77/865877/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/77/865877/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/77/865877/4"}}},"commit":{"parents":[{"commit":"a191d2e629f55c999b37a7a795ad6fbee2807b98","subject":"Merge \"tests: Define a primary key\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a191d2e629f55c999b37a7a795ad6fbee2807b98"}]}],"author":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2022-11-28 17:20:46.000000000","tz":-300},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2022-12-05 16:49:14.000000000","tz":-300},"subject":"Add option for wsrep_sync_wait","message":"Add option for wsrep_sync_wait\n\nWhen using Galera, the wsrep_sync_wait option [1]\ncan change the behavior of a variety of Galera DQL/DML statements\nsuch that a particular operation will pause until outstanding\nwrite-sets are fully persisted to the local node.  The setting\nsupersedes the previous boolean parameter wsrep_causal_reads\nwhich only affected SELECT, with an updated approach that allows\nfor fine-grained control of so-called \"causality checks\"\non individual statement types.  The legacy-compatible setting\nof \u00271\u0027 indicates that READ/SELECT/BEGIN operations should\nproceed only after any pending writesets are fully available.\n\nThe use case for this setting is for an application that\nis running operations on multiple Galera nodes simultaenously.\nAn application that commits data on one node, and then immediately\nuses a different connection (on a potentially different node)\nto SELECT that data, may fail to see those changes if\n\"causality checks\" for SELECT are not enabled.  While\na COMMIT operation in Galera will block locally until all other\nnodes approve of the writeset, the operation does not block\nfor the subsequent period of time when other nodes are actually\npersisting that writeset.  Setting up \"causal reads\"\nin this case indicates that a SELECT operation will wait until\nany writesets in progress are available, thus maintaining\nserialization between the COMMIT and subsequent SELECT.\n\nAs the name implies, wsrep_sync_wait adds...waiting!  to the\noperation, and thus directly impacts performance by adding\nlatency to SELECT operations or to the operations that have\nbeen selected for causality checks, to the degree that\nconcurrent writesets are expected to be present.\n\nSince it\u0027s not expected that most if any Openstack applications\nactually need this setting in order to be effective with\nGalera multi-master operation, and as the setting is available\nwithin client session scope and also impacts performance,\nmaking it available on a per-application basis means that\nspecific applications which may see issues under load can\nchoose to enable this setting, much in the way any other\n\"transaction isolation\" settings might be made, without having\nto add a cluster-wide performance penalty by setting it at the\nGalera server level.\n\n[1] https://mariadb.com/docs/ent/ref/mdb/system-variables/wsrep_sync_wait/\n\nChange-Id: Iee7afcac8ba952a2d67a9ad9dd0e4eae3f42518e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/009d23df45969036c70e4cf59eb4019aaace9a55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/009d23df45969036c70e4cf59eb4019aaace9a55"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
