)]}'
{"id":"openstack%2Foslo.db~865878","triplet_id":"openstack%2Foslo.db~master~I63ad4fbbd7fa72e7a4d8ebd458d323516f997424","project":"openstack/oslo.db","branch":"master","topic":"skip_bandit_on_tests","attention_set":{},"removed_from_attention_set":{"11816":{"account":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"last_update":"2022-12-20 19:23:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I63ad4fbbd7fa72e7a4d8ebd458d323516f997424","subject":"skip bandit on oslo_db/tests","status":"MERGED","created":"2022-11-28 18:31:34.000000000","updated":"2022-12-20 19:24:21.000000000","submitted":"2022-12-20 19:23:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"865878-skip_bandit_on_tests","meta_rev_id":"473f60458d2a81511ad83db7e079af9af5a8daf5","_number":865878,"virtual_id_number":865878,"owner":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-20 19:23:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2022-12-20 17:53:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2022-12-20 17:53:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-28 20:37:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-20 17:53:00.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"41c1387126040f117c5cd062285ac5c7c10c3490","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2022-11-28 18:31:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c49c6da03cc9c1352ed152778c3ce618bcbf634c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 20:37:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a4c068380c84e569c252e243d391fed\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/be4b2cdcf3064f419866ef08af2643da : SUCCESS in 2h 00m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e392b3cde9c4728a272b70ca6ed5f5d : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d23f21228a5c46e19714aa4f6a47ab44 : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a648212fa9044d588b740863b108de1 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42c40fa051f942268d3b3f4240e1d6e1 : SUCCESS in 4m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca971a39a55245efbc9c5549680e0282 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":1},{"id":"240ff8c735b64ff8aa8b38edbf684b7e19fead33","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-12-20 17:53:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c5bcbcc9a75f9d49c59261745ea2bb7d624b1007","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 17:53:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c415e84df10092af70a4b7501c56906545cbd6f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 19:23:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a43e372da0a4c22811019a6b727d474\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a443ba6021ae4f65a29eb0747097e3fc : SUCCESS in 1h 18m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4753f10995e485c945aa52b990eff3c : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff94cdde007941a5a98a4f36b9938a1b : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d599af66af4e45e2b39b82dc3cff6714 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6ff0749ba4c4870ab8a61b64d2d787e : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6313aa7a24bb4bb7bd4ba32a74720aca : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":1},{"id":"3fc47a90b1c8e1ec0d562ea99f4d9c7e3c218638","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 19:23:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"473f60458d2a81511ad83db7e079af9af5a8daf5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 19:24:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ecb499e1c8f46a3829b294c7fcd633b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4145cb4e4c24a52a9acf518bc500707 : SUCCESS in 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77ac27e701f5477fb39b262c1593cd2a : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"43d054403eb697ff96c52ee382365aac1e8d798f","revisions":{"43d054403eb697ff96c52ee382365aac1e8d798f":{"kind":"REWORK","_number":1,"created":"2022-11-28 18:31:34.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/78/865878/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/78/865878/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/78/865878/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/78/865878/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/78/865878/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/78/865878/1"}}},"commit":{"parents":[{"commit":"a191d2e629f55c999b37a7a795ad6fbee2807b98","subject":"Merge \"tests: Define a primary key\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a191d2e629f55c999b37a7a795ad6fbee2807b98"}]}],"author":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2022-11-28 18:30:35.000000000","tz":-300},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2022-11-28 18:30:35.000000000","tz":-300},"subject":"skip bandit on oslo_db/tests","message":"skip bandit on oslo_db/tests\n\nthe bandit command excludes the folder named \"tests\" but does not\nseem to exclude where the tests actually are here, which is\noslo_db/tests.  add this folder, as we have \"assert\" statements\nas well as pickle use in our test suite, which bandit disapproves of.\n\nChange-Id: I63ad4fbbd7fa72e7a4d8ebd458d323516f997424\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/43d054403eb697ff96c52ee382365aac1e8d798f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/43d054403eb697ff96c52ee382365aac1e8d798f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
