)]}'
{"id":"openstack%2Foslo.db~870723","triplet_id":"openstack%2Foslo.db~master~Ifb1dc7ddcb127a69ea01234922caa7ca5ab111ce","project":"openstack/oslo.db","branch":"master","topic":"add_wsrep_sync_wait","attention_set":{},"removed_from_attention_set":{"31857":{"account":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"last_update":"2023-01-18 15:23:45.000000000","reason":"Change was submitted"},"11816":{"account":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"last_update":"2023-01-18 13:16:08.000000000","reason":"\u003cGERRIT_ACCOUNT_11816\u003e replied on the change","reason_account":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}}},"hashtags":[],"change_id":"Ifb1dc7ddcb127a69ea01234922caa7ca5ab111ce","subject":"Fix default value for wsrep_sync_wait option","status":"MERGED","created":"2023-01-17 09:08:26.000000000","updated":"2023-01-18 16:44:16.000000000","submitted":"2023-01-18 15:23:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"870723-add_wsrep_sync_wait","meta_rev_id":"375acc77528144d0f2552680292ad1fa246b2d38","_number":870723,"virtual_id_number":870723,"owner":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-18 15:23:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":2,"date":"2023-01-18 13:16:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":2,"date":"2023-01-18 09:09:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":1,"date":"2023-01-18 13:16:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-17 11:52:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-17 14:20:11.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2023-01-18 09:09:51.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"89557110e7ebf9cc6ea7080666c598a866b89eb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2023-01-17 09:08:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a2a0fd8fab4c88ff563ce40022cca8d544958a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2023-01-17 10:14:13.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"27f969bab32ef73caee0ca8d114c78fe155153ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 11:52:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e501f6ff5814db499fe7167cd2ddb08\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0a53744ab91c4eeeb144128ebc21ae85 : SUCCESS in 1h 35m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9035fd6d2f8470e840bb89665d45aa4 : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61ee396b7c0c4e169f35bd0112952618 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/77a2c6ce27804df8935e7dd905ce0b43 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a1f102a29d445b5b700ae922c2b0046 : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":2},{"id":"8e6be174d020296326c86043427e978968451300","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2023-01-17 13:38:05.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"b58187a6ef6f356c749b9de1a324d362108f3b28","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2023-01-17 13:38:24.000000000","message":"Topic set to add_wsrep_sync_wait","accounts_in_message":[],"_revision_number":3},{"id":"a2bbc1f096c67d13dd42856e4225d299a4850f64","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2023-01-17 14:20:11.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f1523a9450fc6e5635545d963a4105688677b4c9","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2023-01-17 14:24:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bdb5b9b72018d00b9bbcfc01f5a5faa7c823c5f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 15:14:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22438b43e292476f8ac3dfd0cd40537b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ec9099cfe277440baf18bd0ee7f49937 : SUCCESS in 1h 35m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6f89c4959a747cbaaa340da8ba9312d : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82eda9e717a645cb803c3f01ac14716f : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5c5aeaf7d3c54fe8a90f8af2fee37064 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df390a4d86e44d72a54ea12d6029e20f : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":3},{"id":"7e93b0d98e35ba08fb5603d8453e0a3f87fea2c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2023-01-17 15:32:27.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"a38129850add68c8eeef67392d71972dfe9c8b5c","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2023-01-17 15:39:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8e439cbee85643dd59e6a7672e7487f73bfc29fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2023-01-17 16:17:47.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d2189ee241a60d72f16304e6093d129fbddcbaa3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 17:48:53.000000000","message":"Patch Set 5: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf1c4f39d33b410d80f2a4c70656a331\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/18fa92b3b6db4b3cb5ae74e4b4e0ae08 : SUCCESS in 1h 28m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a5d96ce62444cff81aff9e4a6f70dba : FAILURE in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/975917e656b54f10b067916389944d42 : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d5ab2b67df7940e185d3f7ce85ba33f8 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d50148814dfd47a697d0751a7cbfae44 : FAILURE in 3m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb04a85e939b47f3b73116d7a7f7a80a : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":5},{"id":"d739014f7f2f8c76c54672091c89390b96b3b029","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2023-01-18 00:37:59.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"38cf73b0840bf63734fd03c616ba71edb79248b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 01:51:37.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c03b108df54741beb2b65ef98b3cbc27\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4f35ad643d2b492eb588565136a13522 : FAILURE in 1h 06m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb961c15fa704c4da9549b39a0285d7d : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fcbd0115864d45b493661a28530ca406 : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8fde82a87a5949a7a59441dbc5acc862 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b119793b3f074c3180fdb96e9fb2c68c : SUCCESS in 4m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e80cd4b2abfc4e4fb5e940727b275c70 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":6},{"id":"853a2f7ba3a5d55d87628ca77c7a1c8f32613aa6","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2023-01-18 02:16:35.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f91ad44dbf199c3fb2ad8ca8e73b16136ce1f3f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 04:02:45.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/194bbfe73dc34558bc37456b9c0f599d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a91cca10c7164204b1803b0a83265aaf : SUCCESS in 1h 39m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b48c6ce0074c42f7973c5c09dc513f50 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70b6e94525da4164ac2e20b8da0700a5 : SUCCESS in 4m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ae2a55f973444d4ea6ebaeaadf527d3b : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04e43527b761406b819c21a8fb82d562 : SUCCESS in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15672b33eb484e29915712a71c5d0b7c : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":6},{"id":"f97dcb18a4f1c36e4dcd9a5d9d699cf7858456fd","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-01-18 09:09:51.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"facb4a018d848eb3bd4473fbfdc4f5bb49aaf365","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2023-01-18 13:16:08.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"911b1958c1e67aea3f2d4d3c5d5b4a77e55efc48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 13:16:44.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"f4d98b498e867356b4e28ef7b9b7e6fa662b84c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 15:23:45.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8231a4f98cac4ba3994eaa4fac06191b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/336cffe899c44fc6a05fe387d18ec929 : SUCCESS in 1h 42m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17c32a850b374085920cd61c957fa17a : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6eff3d4510c4f86a591ef36b3c4fb4f : SUCCESS in 5m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3f43375b14954431ab42bc5c5c8b8d22 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84ccf3a9c72449fe88bd977037505e37 : SUCCESS in 4m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ebaf3246e17480eb1828dede2579c6d : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":6},{"id":"6ea5dfe5eca6e2f4b8744c9ecd9129303b342cf9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 15:23:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"00e3db1c083c9d5c2ee4e47bc07ee71d2aad54c6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 15:24:55.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3a9bcfba09d423e83b2bd45c4cd0165\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/104647cb19234684b3b128eb1e4a4427 : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9fb91579581d49e79bce6b47036519eb : SUCCESS in 43s","accounts_in_message":[],"_revision_number":6},{"id":"375acc77528144d0f2552680292ad1fa246b2d38","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-01-18 16:44:16.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"b689b6320cdb7dbbb366bf35f23083814f5e79ec","revisions":{"0a3fdfd84abcbfd70ec7a9062f815160e2431942":{"kind":"REWORK","_number":1,"created":"2023-01-17 09:08:26.000000000","uploader":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"ref":"refs/changes/23/870723/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/23/870723/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/23/870723/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/23/870723/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/23/870723/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/23/870723/1"}}},"commit":{"parents":[{"commit":"32b042e7069f90c61548d73e3460ab8e0f6990c7","subject":"Merge \"skip bandit on oslo_db/tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/32b042e7069f90c61548d73e3460ab8e0f6990c7"}]}],"author":{"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","date":"2023-01-17 08:57:20.000000000","tz":0},"committer":{"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","date":"2023-01-17 08:57:20.000000000","tz":0},"subject":"[WIP] fix default value for wsrep_sync_wait option","message":"[WIP] fix default value for wsrep_sync_wait option\n\nThe default value for wsrep_sync_wait option should be `None`.\nHowever, since 0 is set incorrectly, an unintended process is executed.\n\nThis patch fixes default value for wsrep_sync_wait option to `None`\ninstead of `0`.\n\nChange-Id: Ifb1dc7ddcb127a69ea01234922caa7ca5ab111ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/0a3fdfd84abcbfd70ec7a9062f815160e2431942"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/0a3fdfd84abcbfd70ec7a9062f815160e2431942"}]},"branch":"refs/heads/master"},"e4d86aa4e1365bc11c5a6a71bfc839e074ecb7ef":{"kind":"REWORK","_number":2,"created":"2023-01-17 10:14:13.000000000","uploader":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"ref":"refs/changes/23/870723/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/23/870723/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/23/870723/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/23/870723/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/23/870723/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/23/870723/2"}}},"commit":{"parents":[{"commit":"32b042e7069f90c61548d73e3460ab8e0f6990c7","subject":"Merge \"skip bandit on oslo_db/tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/32b042e7069f90c61548d73e3460ab8e0f6990c7"}]}],"author":{"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","date":"2023-01-17 08:57:20.000000000","tz":0},"committer":{"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","date":"2023-01-17 10:14:13.000000000","tz":0},"subject":"[WIP] fix default value for wsrep_sync_wait option","message":"[WIP] fix default value for wsrep_sync_wait option\n\nThe default value for wsrep_sync_wait option should be `None`.\nHowever, since 0 is set incorrectly, an unintended process is executed.\n\nThis patch fixes default value for wsrep_sync_wait option to `None`\ninstead of `0`.\n\nChange-Id: Ifb1dc7ddcb127a69ea01234922caa7ca5ab111ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e4d86aa4e1365bc11c5a6a71bfc839e074ecb7ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e4d86aa4e1365bc11c5a6a71bfc839e074ecb7ef"}]},"branch":"refs/heads/master"},"bee6c77dd119429be9f681dd72a84511d7ff4e85":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-01-17 13:38:05.000000000","uploader":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"ref":"refs/changes/23/870723/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/23/870723/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/23/870723/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/23/870723/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/23/870723/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/23/870723/3"}}},"commit":{"parents":[{"commit":"32b042e7069f90c61548d73e3460ab8e0f6990c7","subject":"Merge \"skip bandit on oslo_db/tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/32b042e7069f90c61548d73e3460ab8e0f6990c7"}]}],"author":{"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","date":"2023-01-17 08:57:20.000000000","tz":0},"committer":{"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","date":"2023-01-17 13:38:05.000000000","tz":0},"subject":"fix default value for wsrep_sync_wait option","message":"fix default value for wsrep_sync_wait option\n\nThe default value for wsrep_sync_wait option should be `None`.\nHowever, since 0 is set incorrectly, an unintended process is executed.\n\nThis patch fixes default value for wsrep_sync_wait option to `None`\ninstead of `0`.\n\nChange-Id: Ifb1dc7ddcb127a69ea01234922caa7ca5ab111ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/bee6c77dd119429be9f681dd72a84511d7ff4e85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/bee6c77dd119429be9f681dd72a84511d7ff4e85"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8e2c431d2b44f53253cb64a4393c6290fa677628":{"kind":"REWORK","_number":4,"created":"2023-01-17 15:32:27.000000000","uploader":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"ref":"refs/changes/23/870723/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/23/870723/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/23/870723/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/23/870723/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/23/870723/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/23/870723/4"}}},"commit":{"parents":[{"commit":"32b042e7069f90c61548d73e3460ab8e0f6990c7","subject":"Merge \"skip bandit on oslo_db/tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/32b042e7069f90c61548d73e3460ab8e0f6990c7"}]}],"author":{"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","date":"2023-01-17 08:57:20.000000000","tz":0},"committer":{"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","date":"2023-01-17 15:31:56.000000000","tz":0},"subject":"Fix default value for wsrep_sync_wait option","message":"Fix default value for wsrep_sync_wait option\n\nThe default value for wsrep_sync_wait option should be `None`.\nHowever, since 0 is set incorrectly, an unintended process is executed.\n\nThis patch fixes default value for wsrep_sync_wait option to `None`\ninstead of `0`.\n\nChange-Id: Ifb1dc7ddcb127a69ea01234922caa7ca5ab111ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/8e2c431d2b44f53253cb64a4393c6290fa677628"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/8e2c431d2b44f53253cb64a4393c6290fa677628"}]},"branch":"refs/heads/master"},"0ebe16ec3b78e40d1380ba4f75bb88643699d53e":{"kind":"REWORK","_number":5,"created":"2023-01-17 16:17:47.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/23/870723/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/23/870723/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/23/870723/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/23/870723/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/23/870723/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/23/870723/5"}}},"commit":{"parents":[{"commit":"32b042e7069f90c61548d73e3460ab8e0f6990c7","subject":"Merge \"skip bandit on oslo_db/tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/32b042e7069f90c61548d73e3460ab8e0f6990c7"}]}],"author":{"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","date":"2023-01-17 08:57:20.000000000","tz":0},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2023-01-17 16:17:15.000000000","tz":-300},"subject":"Fix default value for wsrep_sync_wait option","message":"Fix default value for wsrep_sync_wait option\n\nThe default value for wsrep_sync_wait option should be `None`.\nHowever, since 0 is set incorrectly, an unintended process is executed.\n\nThis patch fixes default value for wsrep_sync_wait option to `None`\ninstead of `0`.\n\nChange-Id: Ifb1dc7ddcb127a69ea01234922caa7ca5ab111ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/0ebe16ec3b78e40d1380ba4f75bb88643699d53e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/0ebe16ec3b78e40d1380ba4f75bb88643699d53e"}]},"branch":"refs/heads/master"},"b689b6320cdb7dbbb366bf35f23083814f5e79ec":{"kind":"REWORK","_number":6,"created":"2023-01-18 00:37:59.000000000","uploader":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"ref":"refs/changes/23/870723/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/23/870723/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/23/870723/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/23/870723/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/23/870723/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/23/870723/6"}}},"commit":{"parents":[{"commit":"32b042e7069f90c61548d73e3460ab8e0f6990c7","subject":"Merge \"skip bandit on oslo_db/tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/32b042e7069f90c61548d73e3460ab8e0f6990c7"}]}],"author":{"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","date":"2023-01-17 08:57:20.000000000","tz":0},"committer":{"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","date":"2023-01-18 00:31:22.000000000","tz":0},"subject":"Fix default value for wsrep_sync_wait option","message":"Fix default value for wsrep_sync_wait option\n\nThe default value for wsrep_sync_wait option should be `None`.\nHowever, since 0 is set incorrectly, an unintended process is executed.\n\nThis patch fixes default value for wsrep_sync_wait option to `None`\ninstead of `0`.\n\nChange-Id: Ifb1dc7ddcb127a69ea01234922caa7ca5ab111ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/b689b6320cdb7dbbb366bf35f23083814f5e79ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/b689b6320cdb7dbbb366bf35f23083814f5e79ec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
