)]}'
{"id":"openstack%2Foslo.db~874858","triplet_id":"openstack%2Foslo.db~master~Iba29ded227634e02795052acfd89b572bf21f54c","project":"openstack/oslo.db","branch":"master","topic":"bug/2008209","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-03-20 19:07:31.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-02-23 16:04:23.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"11816":{"account":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"last_update":"2023-03-20 13:23:29.000000000","reason":"\u003cGERRIT_ACCOUNT_11816\u003e replied on the change","reason_account":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}}},"hashtags":[],"change_id":"Iba29ded227634e02795052acfd89b572bf21f54c","subject":"Rollback the connection after server ping method","status":"MERGED","created":"2023-02-23 09:22:33.000000000","updated":"2023-03-20 19:08:30.000000000","submitted":"2023-03-20 19:07:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874858-bug/2008209","meta_rev_id":"09742376a5336209ef6a4b00ef92973dc34e4540","_number":874858,"virtual_id_number":874858,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-20 19:07:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2023-03-20 13:23:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":1,"date":"2023-03-07 00:26:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2023-03-10 23:02:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2023-03-20 13:23:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-23 09:22:55.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2023-02-23 09:22:55.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2023-02-23 11:22:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-02 12:24:28.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2023-03-07 00:26:44.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-03-10 23:02:00.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"8c07bc38cae1ae8d17d8bc3825b48567ef1424d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-23 09:22:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1a97411e22b73ad1e7b9ad05ad8369d25e5603b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-02-23 11:14:32.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"cbfbceaceb22110d3ffe2a453c09b294335129a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 11:22:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d786ea27f1864f429ca6ee472a39a93c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7ea4535411594f2eb31945395109362c : SUCCESS in 1h 32m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1664d883ad34400c954e116a79b4fb5d : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/823de74a721941cabcde83314b5bbd35 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/785f3cfd18dc41ff9bd556e7019a61a9 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f54e45927c4e43bc926c067c32bc4f8f : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":1},{"id":"fcb99c76d6ab93047edccfc078e5f38347c5d445","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-23 14:56:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"33da9fc1198cc4295f9a6ee66aa848140273df89","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-23 14:56:19.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"fb7aaada3b8a6f1f70e802a918ed9dc493a22d1f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-02-23 16:04:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"92d4219470bef7d9848b3cf857d41aec51b0c1bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 16:32:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb445928240d46c9b79b64f13248f409\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/83451a2d7b554917b956e093a7c66b49 : SUCCESS in 1h 32m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4965c63c334d4da9b87441764ed3e70a : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3f64448943343738d8c17247fdf6286 : FAILURE in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/be3c9b7da000407c86257d364693d36c : FAILURE in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4569eecc03a3471e948272baac72cb20 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":2},{"id":"d9a491669a4b2e747330f6931cdbccb246936ab5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-23 16:54:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"42cc20e3e20dad170f99249903070ff63c9c0342","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-02-23 16:54:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"30ccb3001678e0eb686f293acfcfdc6972f92a09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 18:19:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca43ccdd19004799b3a3622df6833f04\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9f5b65c1c46e44d18174efee0df077cb : SUCCESS in 1h 24m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a09877eb97f24cac98944559a6347753 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b962a69b0124bbb9530a5dbc3c406cb : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b96b3facdda7499d84aeb693bf1cb4f6 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b4db39575c84e5587ca5f107fc11d61 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":3},{"id":"98ff9a71e2798153fd75527d24fc5fa84b805f2b","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2023-02-24 19:18:02.000000000","message":"Patch Set 3: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"30de1371174700b3f23b0619eb319402edddda01","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-02 09:05:08.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"2db2d084d2c0f71c9b0985ac9548294f19ee920d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-02 09:05:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a98c7dfbe1b049827b9e7a2acf1da817c61247a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 10:43:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3dfb5bc9239f49aa8f91c05a5efca083\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c3ce430be8194218b2bb52382ca123cd : SUCCESS in 1h 36m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6711ff1a2004ca2859bcbfc867a968c : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4917d85332704d9caa591f3f63d95033 : SUCCESS in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cc17762ab59a49dd9aeaed85045e22fe : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93626310a3ef431283e1489615ae4710 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":4},{"id":"48c0ae5d19f763e9bbb90c42678a7046004a73f6","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-02 12:24:28.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"6e57a8d95894b26c21d4915816c3a611b6d26e7e","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2023-03-02 12:58:08.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2d095d7d6b7c0248ab11e5fe6cd4d236603de854","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-02 16:52:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"405292e7fa85c8dd417e157d570409c19cd70247","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-02 16:52:57.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"d40e829c62c3d5abd51acaaae977850bad2d18b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 18:35:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55ee0c7872ca482a9d34ab3d0b88171f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/27053183a0aa42ff8a3c3d3a7c2b1b88 : SUCCESS in 1h 36m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b4c64f8fc0d4d37b2016b1ad03ece0f : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e755f277b05c44fd933eca0c42c22f7e : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01841798ff0e41fdae7a5cac8c01b6c9 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a819d261f6e412e86f80c4b95da7cc8 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":5},{"id":"50cd53b4ae2561749deadb1007875b9921affaa3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-03-07 00:26:44.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9c0c87653f7246ae8c440c289e2abb138476c724","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-03-10 23:02:00.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"5f991f1dd84603ab4f7b0faa5782d6d350f69083","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2023-03-20 13:23:29.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"dd98700e0275ceba6bcffab190ab90a75e37ffd3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 13:23:58.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a483cbbbd7b8685b0faa493809db28d2704d58b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 14:14:05.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/77413e315a4042328d8030864bffdfd5\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d721079551154ceb990e335eafd00601 : FAILURE in 44m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e28c1cce3a56474a9a41b99af10f7d13 : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33c1fc02bb964c158b36d489e66dea34 : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/67473499411441e1a9ec949a90a0fc94 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78c4e808c0e24cf7891d1a879025c3f4 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":5},{"id":"c51428c9f23a98f928a29f33ecf7b04b1f3acbb5","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-03-20 15:50:36.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"025258d340dea03bb7200020c1f2bf6c665945a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 17:29:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/451e7dc03d36463ea80c4f9f29aa5f6c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/afa5f2640e804f0cbd985aed0bd192b8 : SUCCESS in 1h 30m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3be0cb36671546d9b4247b8c6dbb6784 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc3e2ff4c8f34a838c584944b74a884e : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d2a610d88be84c2c84cd516314848011 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b84ef568aaf44d8aa1641917620f9f48 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":5},{"id":"5b7db725d0fced1d6b186b15498e8a94eefcd4ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 17:29:31.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"5e0d98c960ae09765219f22f47a0c5bdf4741347","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 19:07:30.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc8d4237107a4d0e8f3773a05d3fde26\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/505f1675c44e44979d0b4d69be4965a8 : SUCCESS in 1h 33m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adcbc04b94ac44ab8847dc93537c3d98 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71334ca06b074cc8b894e91e4cd536c1 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e59839a0c69a4c0f95c6b6553cf6e18e : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12d472418ce74b97b045f496bf26f552 : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":5},{"id":"435da9d2801a82b52f64460137031dfd6b2cc80c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 19:07:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"09742376a5336209ef6a4b00ef92973dc34e4540","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 19:08:30.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d37714b09d89403f820727e9b088d057\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/100da6235cf142b498dac0d5f8d91c2d : SUCCESS in 40s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"877bcfc6a6ed16ba6885f47824df6b1f5ac60b4e","revisions":{"3d8a63254da3a9a259e2546b86f1474b3a28270c":{"kind":"REWORK","_number":1,"created":"2023-02-23 09:22:33.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/58/874858/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/58/874858/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/58/874858/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/58/874858/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/58/874858/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/58/874858/1"}}},"commit":{"parents":[{"commit":"7d619ae3eb2a64dadeaa9fbaeb1b9424b83414a6","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7d619ae3eb2a64dadeaa9fbaeb1b9424b83414a6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-18 12:52:22.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-18 12:54:23.000000000","tz":60},"subject":"Rollback the connection after server ping method","message":"Rollback the connection after server ping method\n\nIn the method ``engines._connect_ping_listener``, the connection should\nbe rolled back after the ping execution. The rollback will revert the\ntransaction and delete it.\n\nCloses-Bug: #2008209\nChange-Id: Iba29ded227634e02795052acfd89b572bf21f54c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/3d8a63254da3a9a259e2546b86f1474b3a28270c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/3d8a63254da3a9a259e2546b86f1474b3a28270c"}]},"branch":"refs/heads/master"},"05bcc2b78b1d0f58f2249adc69a3e2296f1fd54c":{"kind":"REWORK","_number":2,"created":"2023-02-23 14:56:06.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/58/874858/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/58/874858/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/58/874858/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/58/874858/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/58/874858/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/58/874858/2"}}},"commit":{"parents":[{"commit":"7d619ae3eb2a64dadeaa9fbaeb1b9424b83414a6","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7d619ae3eb2a64dadeaa9fbaeb1b9424b83414a6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-18 12:52:22.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-18 18:28:18.000000000","tz":60},"subject":"Rollback the connection after server ping method","message":"Rollback the connection after server ping method\n\nIn the method ``engines._connect_ping_listener``, the connection should\nbe rolled back after the ping execution. The rollback will revert the\ntransaction and delete it.\n\nCloses-Bug: #2008209\nChange-Id: Iba29ded227634e02795052acfd89b572bf21f54c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/05bcc2b78b1d0f58f2249adc69a3e2296f1fd54c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/05bcc2b78b1d0f58f2249adc69a3e2296f1fd54c"}]},"branch":"refs/heads/master"},"e95ea97b75d33bc4869532177152c52e712d267e":{"kind":"REWORK","_number":3,"created":"2023-02-23 16:54:11.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/58/874858/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/58/874858/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/58/874858/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/58/874858/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/58/874858/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/58/874858/3"}}},"commit":{"parents":[{"commit":"7d619ae3eb2a64dadeaa9fbaeb1b9424b83414a6","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7d619ae3eb2a64dadeaa9fbaeb1b9424b83414a6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-18 12:52:22.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-18 20:26:23.000000000","tz":60},"subject":"Rollback the connection after server ping method","message":"Rollback the connection after server ping method\n\nIn the method ``engines._connect_ping_listener``, the connection should\nbe rolled back after the ping execution. The rollback will revert the\ntransaction and delete it.\n\nCloses-Bug: #2008209\nChange-Id: Iba29ded227634e02795052acfd89b572bf21f54c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e95ea97b75d33bc4869532177152c52e712d267e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e95ea97b75d33bc4869532177152c52e712d267e"}]},"branch":"refs/heads/master"},"349393dd8414736c86ebe683ce4504d34fc6c335":{"kind":"REWORK","_number":4,"created":"2023-03-02 09:05:37.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/58/874858/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/58/874858/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/58/874858/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/58/874858/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/58/874858/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/58/874858/4"}}},"commit":{"parents":[{"commit":"5cd7962a2d35902703aac3b68b85e50e26ff4981","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/5cd7962a2d35902703aac3b68b85e50e26ff4981"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-18 12:52:22.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-01 18:04:31.000000000","tz":60},"subject":"Rollback the connection after server ping method","message":"Rollback the connection after server ping method\n\nIn the method ``engines._connect_ping_listener``, the connection should\nbe rolled back after the ping execution. The rollback will revert the\ntransaction and delete it.\n\nCloses-Bug: #2008209\nChange-Id: Iba29ded227634e02795052acfd89b572bf21f54c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/349393dd8414736c86ebe683ce4504d34fc6c335"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/349393dd8414736c86ebe683ce4504d34fc6c335"}]},"branch":"refs/heads/master"},"877bcfc6a6ed16ba6885f47824df6b1f5ac60b4e":{"kind":"REWORK","_number":5,"created":"2023-03-02 16:52:57.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/58/874858/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/58/874858/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/58/874858/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/58/874858/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/58/874858/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/58/874858/5"}}},"commit":{"parents":[{"commit":"5cd7962a2d35902703aac3b68b85e50e26ff4981","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/5cd7962a2d35902703aac3b68b85e50e26ff4981"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-02-18 12:52:22.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-03-02 16:52:32.000000000","tz":0},"subject":"Rollback the connection after server ping method","message":"Rollback the connection after server ping method\n\nIn the method ``engines._connect_ping_listener``, the connection should\nbe rolled back after the ping execution. The rollback will revert the\ntransaction and delete it.\n\nCloses-Bug: #2008209\nChange-Id: Iba29ded227634e02795052acfd89b572bf21f54c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/877bcfc6a6ed16ba6885f47824df6b1f5ac60b4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/877bcfc6a6ed16ba6885f47824df6b1f5ac60b4e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
