)]}'
{"id":"openstack%2Foslo.db~874935","triplet_id":"openstack%2Foslo.db~master~Ibf46f015271497221c1c3d5655b2c2bf799d1f3d","project":"openstack/oslo.db","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-03-03 17:34:49.000000000","reason":"Change was abandoned"},"11816":{"account":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"last_update":"2023-03-03 17:34:49.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ibf46f015271497221c1c3d5655b2c2bf799d1f3d","subject":"Get 2.0 to pass fully with test suite","status":"ABANDONED","created":"2023-02-23 16:07:19.000000000","updated":"2023-03-03 17:34:49.000000000","total_comment_count":11,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"108e8b588b53a59ea32e445e0dbed85fa168f01d","_number":874935,"virtual_id_number":874935,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-03-01 22:17:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":-2,"date":"2023-03-01 20:47:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"REVIEWER":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-23 17:10:54.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2023-02-23 17:32:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-24 19:55:03.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"baf999552f5a461c33fd9332f464e3cf3b564ae2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-02-23 16:07:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1df0dac0ba738a4acf151df3146218c513ff37d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 17:32:23.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f94c4ee8d03544c5aa9688b8c67ead93\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8ce93777e80949edbf50107d21895a8e : SUCCESS in 1h 21m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88a21403bbb34561953b489b6d002410 : FAILURE in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b05749f5534d490ca48e7b55cd689e7d : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c7862610705046c9b34853c7229a82fb : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d836415c56448e3bd82669a3307a4f1 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":1},{"id":"3552afedb288047779bcb19a40737674696cc897","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-02-23 19:26:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"eba4c03712ca71395da70d330762fc4e6cfd99ee","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-02-23 19:27:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"951df67ceb630f91425a9be5a1fa861fedfe66c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 21:00:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f2075ed38dd45b3b0c103a61eb9b7cb\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/01204d7aa8b8455bb983a775636d3a7a : SUCCESS in 1h 33m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52403cd978ed4eed91c11370eeb2100b : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1fbbf90fdf54b8b91ccef8d138c270e : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b620b80e954d4913804b949aab54aa47 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d3658e857f14d72bd2d101c285af9f0 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":2},{"id":"ef0b09d34f44b3d80066a632071009924bce1bb2","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2023-02-24 19:55:03.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"96dda037765b4c92a2d63dc9e9cae8a9db5de22a","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2023-02-24 20:05:46.000000000","message":"Patch Set 2: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"639804d11dfc651358a79000dba1615c42958ea2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2023-02-24 20:09:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"852c509bca4d751578105a707418945552ed6616","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 21:34:54.000000000","message":"Patch Set 3: Verified-1\n\n(12 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b04b4ec37cd43a7aac85b1ebd233fb3\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e064bbf1fcdd4eccbaf655e525736388 : SUCCESS in 1h 20m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2bce6322693403e8d82082d22c8586b : FAILURE in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6777451c278d4f4db4ba92c0c0c9ed75 : FAILURE in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a493af279b234b90a3c14a2ead76f03e : FAILURE in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/880d820ceea14e58a9d4bd3aec24e050 : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":3},{"id":"8e80804d288d063836d0ed739b97772ed4787f19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2023-02-24 21:53:01.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"09d9bb41a9d95ac21cc97b0135c5f4b753394fdd","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2023-02-24 22:01:10.000000000","message":"Patch Set 4: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6cdd8493d9a14aabe02c225e3e5d3d6ed791be03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 23:13:34.000000000","message":"Patch Set 4: Verified-1\n\n(17 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b6ccf4b2628455cae4acc0a0233ff52\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c313c99434d64840bf58ef38c40fede2 : SUCCESS in 1h 19m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/849a194d75be445d8c7aee9f3fa2b889 : FAILURE in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f6ca2f008c84ab38caab983bda12a44 : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e26a4e6865b64e67abb37ea123a32a09 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7addb7b1a266446a98fd2820a2f62a0a : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":4},{"id":"c5f69eda12025d3f0ef31df1487134f9dc4c10ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2023-02-26 16:29:12.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a3b1c2c0908dbcba32e5fa3ea48758014dcded8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-26 18:21:00.000000000","message":"Patch Set 5: Verified-1\n\n(25 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/19e6cfca9dac4c76a4071b60451e0973\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/62d96143d6e64c1d97bf1df6e46f0a42 : SUCCESS in 1h 45m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8ecc95061544faca2dcd05b28192a84 : FAILURE in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9288484ac0ff44999c2064e1633d4dda : FAILURE in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3b1b40b43ffa4125b915945125bd71fb : FAILURE in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab69a91cb7c341aeb6a26c3b987d8b0d : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":5},{"id":"3275d07a4e976bd0a33489874e50ed788e874ec9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-02-27 14:47:26.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2f568afc22fbc80ad9247ba1b1dbe2cd71a01bf8","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2023-02-27 14:50:59.000000000","message":"Patch Set 5: Code-Review-2\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"d1037df9fde728d3a9a849f78add251656dd1263","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-02-28 10:20:26.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"c0df372cb2b1e3fe8d09b1e8ce86c3f5c00b008c","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2023-02-28 13:57:17.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e8956e19d328b2a4d791106ba1dbd5ad2befe5c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2023-03-01 20:47:33.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f2091a1aafbbfbade9367a9d8ae5c223950159b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 22:17:22.000000000","message":"Patch Set 6: Verified-1\n\n(39 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f6d070abba21443aa78ca10c6bdbdfeb\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5467d8ac4f354ebdb00d66e8ddec200f : SUCCESS in 1h 27m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c905e7b10854281a792f02911b53e51 : FAILURE in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b9258b8b5da4403a0eaa432861b2020 : FAILURE in 6m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cfc1ff7d9cb547008847899c601cffb2 : FAILURE in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56827d5ae0074788bf7733208027ab8b : SUCCESS in 5m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac1c187377d24c6da5c03b754f511dfd : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":6},{"id":"108e8b588b53a59ea32e445e0dbed85fa168f01d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2023-03-03 17:34:49.000000000","message":"Abandoned\n\nIfaca67c07f008d8bc0febeecd3e200cc7ee7a4b0 is coming along here and I will look to re-propose the changes from a series of patches on that end.   pre-ping support depends on SQLAlchemy 2.0.5 which IMO should depend on removing sqlalchemy-migrate","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"4693cadafccd9cd2473e0787672206529cee3863","revisions":{"1bfa01c288307830d71590d20cc3c318d25e9580":{"kind":"REWORK","_number":1,"created":"2023-02-23 16:07:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/35/874935/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/35/874935/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/35/874935/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/35/874935/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/35/874935/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/35/874935/1"}}},"commit":{"parents":[{"commit":"7d619ae3eb2a64dadeaa9fbaeb1b9424b83414a6","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7d619ae3eb2a64dadeaa9fbaeb1b9424b83414a6"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-02-23 15:03:36.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-02-23 15:03:36.000000000","tz":0},"subject":"Use \u0027pool_pre_ping\u0027","message":"Use \u0027pool_pre_ping\u0027\n\nThis functionality has been available upstream since SQLALchemy 1.2 [1].\nUse it.\n\n[1] https://docs.sqlalchemy.org/en/20/core/engines.html#sqlalchemy.create_engine.params.pool_pre_ping\n\nChange-Id: Ibf46f015271497221c1c3d5655b2c2bf799d1f3d\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/1bfa01c288307830d71590d20cc3c318d25e9580"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/1bfa01c288307830d71590d20cc3c318d25e9580"}]},"branch":"refs/heads/master"},"34b56e3256cfdf35b9e8cbb904506d6c8c50f0c3":{"kind":"REWORK","_number":2,"created":"2023-02-23 19:26:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/35/874935/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/35/874935/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/35/874935/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/35/874935/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/35/874935/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/35/874935/2"}}},"commit":{"parents":[{"commit":"7d619ae3eb2a64dadeaa9fbaeb1b9424b83414a6","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7d619ae3eb2a64dadeaa9fbaeb1b9424b83414a6"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-02-23 15:03:36.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-02-23 15:03:36.000000000","tz":0},"subject":"Use \u0027pool_pre_ping\u0027","message":"Use \u0027pool_pre_ping\u0027\n\nThis functionality has been available upstream since SQLALchemy 1.2 [1].\nUse it.\n\n[1] https://docs.sqlalchemy.org/en/20/core/engines.html#sqlalchemy.create_engine.params.pool_pre_ping\n\nChange-Id: Ibf46f015271497221c1c3d5655b2c2bf799d1f3d\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/34b56e3256cfdf35b9e8cbb904506d6c8c50f0c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/34b56e3256cfdf35b9e8cbb904506d6c8c50f0c3"}]},"branch":"refs/heads/master"},"3c92e4aa3d08138922624132e8d25a3899950d71":{"kind":"REWORK","_number":3,"created":"2023-02-24 20:09:43.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/35/874935/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/35/874935/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/35/874935/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/35/874935/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/35/874935/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/35/874935/3"}}},"commit":{"parents":[{"commit":"7d619ae3eb2a64dadeaa9fbaeb1b9424b83414a6","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7d619ae3eb2a64dadeaa9fbaeb1b9424b83414a6"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-02-23 15:03:36.000000000","tz":0},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2023-02-24 20:08:49.000000000","tz":-300},"subject":"Use \u0027pool_pre_ping\u0027","message":"Use \u0027pool_pre_ping\u0027\n\nThis functionality has been available upstream since SQLALchemy 1.2 [1].\nUse it.\n\nHowever, we need to have event support within ping listeners,\nwhich is only available as of SQLAlchemy 2.0 and is broken for\nMySQL / PostgreSQL until [2] is fixed.\n\n[1] https://docs.sqlalchemy.org/en/20/core/engines.html#sqlalchemy.create_engine.params.pool_pre_ping\n[2] https://github.com/sqlalchemy/sqlalchemy/issues/5648\n\nChange-Id: Ibf46f015271497221c1c3d5655b2c2bf799d1f3d\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/3c92e4aa3d08138922624132e8d25a3899950d71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/3c92e4aa3d08138922624132e8d25a3899950d71"}]},"branch":"refs/heads/master"},"7bd677fd559c8c215fc7bb0d488ea88aca36bc2f":{"kind":"REWORK","_number":4,"created":"2023-02-24 21:53:01.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/35/874935/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/35/874935/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/35/874935/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/35/874935/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/35/874935/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/35/874935/4"}}},"commit":{"parents":[{"commit":"7d619ae3eb2a64dadeaa9fbaeb1b9424b83414a6","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7d619ae3eb2a64dadeaa9fbaeb1b9424b83414a6"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-02-23 15:03:36.000000000","tz":0},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2023-02-24 21:51:47.000000000","tz":-300},"subject":"Use \u0027pool_pre_ping\u0027","message":"Use \u0027pool_pre_ping\u0027\n\nWIP: there are many elements of oslo.db that do not work with\nSQLAlchemy 2.0 still.  As I\u0027m running against 2.0 for real here,\nim trying to work through these incompatibilities; we will definitely\nneed a compat module again.\n\nThis functionality has been available upstream since SQLALchemy 1.2 [1].\nUse it.\n\nHowever, we need to have event support within ping listeners,\nwhich is only available as of SQLAlchemy 2.0 and is broken for\nMySQL / PostgreSQL until [2] is fixed.\n\n[1] https://docs.sqlalchemy.org/en/20/core/engines.html#sqlalchemy.create_engine.params.pool_pre_ping\n[2] https://github.com/sqlalchemy/sqlalchemy/issues/5648\n\nChange-Id: Ibf46f015271497221c1c3d5655b2c2bf799d1f3d\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7bd677fd559c8c215fc7bb0d488ea88aca36bc2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7bd677fd559c8c215fc7bb0d488ea88aca36bc2f"}]},"branch":"refs/heads/master"},"b04ae16d5727474ac688e036689d8f16873bcf44":{"kind":"REWORK","_number":5,"created":"2023-02-26 16:29:12.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/35/874935/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/35/874935/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/35/874935/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/35/874935/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/35/874935/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/35/874935/5"}}},"commit":{"parents":[{"commit":"7d619ae3eb2a64dadeaa9fbaeb1b9424b83414a6","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7d619ae3eb2a64dadeaa9fbaeb1b9424b83414a6"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-02-23 15:03:36.000000000","tz":0},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2023-02-26 16:22:56.000000000","tz":-300},"subject":"Use \u0027pool_pre_ping\u0027","message":"Use \u0027pool_pre_ping\u0027\n\nWIP: there are many elements of oslo.db that do not work with\nSQLAlchemy 2.0 still.  As I\u0027m running against 2.0 for real here,\nim trying to work through these incompatibilities; we will definitely\nneed a compat module again:\n\n* never call str() on a URL and expect it to be meaningful anymore.\n  The password is aggressively obfuscated now (users absolultely\n  wouldn\u0027t let us leave it as is)\n* more utilities and fixtures that were calling begin() within a\n  block that would have already begun\n* isnot is now called is_not; mocking \"isnot\" leads into too many\n  weird compat layers\n* ORM InstrumentedAttribute and internals use __slots__ now, mock\n  seems to not be able to patch methods.  ideally these tests would use\n  a comparator subclass or something\n* Connection.connection.connection is now called driver_connection,\n  SQLAlchemy keeps the old name available however oslo.db test suite\n  does not appear to tolerate the deprecation warning emitted,\n  so add a compat layer\n* mapper() is fully removed from 2.0, not sure if there is another\n  not-yet-committed gerrit that removes mapper()\n* since we want to use handle_error support within\n  ping listeners (new in 2.0), we need to receive the handle_error event\n  without assuming \"engine\" is present, and use the new \"dialect\" attribute\n  to get at the dialect.\n\nFor ping listener:\n\nThis functionality has been available upstream since SQLALchemy 1.2 [1].\nUse it; however, we need to have event support within ping listeners,\nwhich is only available as of SQLAlchemy 2.0 and is broken for\nMySQL / PostgreSQL until [2] is fixed.\n\n[1] https://docs.sqlalchemy.org/en/20/core/engines.html#sqlalchemy.create_engine.params.pool_pre_ping\n[2] https://github.com/sqlalchemy/sqlalchemy/issues/5648\n\nChange-Id: Ibf46f015271497221c1c3d5655b2c2bf799d1f3d\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/b04ae16d5727474ac688e036689d8f16873bcf44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/b04ae16d5727474ac688e036689d8f16873bcf44"}]},"branch":"refs/heads/master"},"4693cadafccd9cd2473e0787672206529cee3863":{"kind":"REWORK","_number":6,"created":"2023-03-01 20:47:33.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/35/874935/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/35/874935/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/35/874935/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/35/874935/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/35/874935/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/35/874935/6"}}},"commit":{"parents":[{"commit":"7b43ae34c729f70f7c99fad566eeff553019140f","subject":"Remove sqlalchemy-migrate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7b43ae34c729f70f7c99fad566eeff553019140f"}]}],"author":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2023-02-23 15:03:36.000000000","tz":0},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2023-03-01 20:46:57.000000000","tz":-300},"subject":"Get 2.0 to pass fully with test suite","message":"Get 2.0 to pass fully with test suite\n\nWe\u0027d like to use pool_pre_ping, added in SQLAlchemy at\n[1].  However, oslo.db needs to intercept disconnect errors\nwithin ping to work correctly (like to handle special Galera states\netc), so we need a new 2.0 feature described at [2].   Also, 2.0 didn\u0027t\nimplement it correctly, so we need 2.0.5 for it to really\nwork [3].\n\nThis patch adds \"native pre ping\" as an experimental option\nfor the moment just tested internally.  the old ping handler\nis tested as well.\n\nFor now, get everything 2.0 working just as a test so we can clean\nup later.\n\nOtherwise, we have some issues w the test suite:\n\n* never call str() on a URL and expect it to be meaningful anymore.\n  The password is aggressively obfuscated now (users absolultely\n  wouldn\u0027t let us leave it as is)\n* more utilities and fixtures that were calling begin() within a\n  block that would have already begun\n* isnot is now called is_not; mocking \"isnot\" leads into too many\n  weird compat layers\n* ORM InstrumentedAttribute and internals use __slots__ now, mock\n  seems to not be able to patch methods.  ideally these tests would use\n  a comparator subclass or something\n* Connection.connection.connection is now called driver_connection,\n  SQLAlchemy keeps the old name available however oslo.db test suite\n  does not appear to tolerate the deprecation warning emitted,\n  so add a compat layer\n* mapper() is fully removed from 2.0, not sure if there is another\n  not-yet-committed gerrit that removes mapper()\n* since we want to use handle_error support within\n  ping listeners (new in 2.0), we need to receive the handle_error event\n  without assuming \"engine\" is present, and use the new \"dialect\" attribute\n  to get at the dialect.\n\nFor ping listener:\n\nThis functionality has been available upstream since SQLALchemy 1.2 [1].\nUse it; however, we need to have event support within ping listeners,\nwhich is only available as of SQLAlchemy 2.0 and is broken for\nMySQL / PostgreSQL until [2] is fixed.\n\n[1] https://docs.sqlalchemy.org/en/20/core/engines.html#sqlalchemy.create_engine.params.pool_pre_ping\n[2] https://docs.sqlalchemy.org/en/20/changelog/changelog_20.html#change-2fe37eaf2295cebd3bb4ee8e5b8c575c\n[3] https://github.com/sqlalchemy/sqlalchemy/issues/5648\n\nChange-Id: Ibf46f015271497221c1c3d5655b2c2bf799d1f3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/4693cadafccd9cd2473e0787672206529cee3863"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/4693cadafccd9cd2473e0787672206529cee3863"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
