)]}'
{"id":"openstack%2Foslo.db~875627","triplet_id":"openstack%2Foslo.db~master~I3008562db52a5625504b262d7148528a832246dd","project":"openstack/oslo.db","branch":"master","topic":"bug/2008720","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-06-23 09:45:00.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"11816":{"account":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"last_update":"2023-04-17 09:40:19.000000000","reason":"Change was abandoned"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-04-17 09:40:19.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3008562db52a5625504b262d7148528a832246dd","subject":"Ignore unset opts when deprecated","status":"MERGED","created":"2023-02-28 10:21:23.000000000","updated":"2023-06-23 13:36:47.000000000","submitted":"2023-06-23 13:35:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"875627-bug/2008720","meta_rev_id":"135e7db8a39d7239eb010883aaac038746cd5f6d","_number":875627,"virtual_id_number":875627,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-23 13:35:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2023-06-02 13:17:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2023-06-13 21:03:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-28 10:33:38.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2023-02-28 10:50:38.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2023-02-28 11:50:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-28 16:26:58.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-02-28 16:26:58.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2023-04-17 12:26:14.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"b304f32c0d76bb9779769f7308686cc3b171a022","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-02-28 10:21:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a1f80c89309edec8e40f858a6c924035a1e069f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2023-02-28 10:33:38.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aea53234335c5dd7e2f7a89caffc9c195c58e845","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-02-28 10:50:38.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8815b49418f5a11a306ee7cd64eb876f7a213a41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 11:50:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c4926852eda48be817ed7654c682d8f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3e8ae35ce03f4573a270dfa3f5a93590 : SUCCESS in 1h 23m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c36c4446d3e2436dbe42d64122b8d8c5 : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/896bc497dead46d9b6e3c6d0fe6f04c7 : FAILURE in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe06240b0f624ab7bc9a8c741d10c38c : FAILURE in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ef522dde95249bfa73e228135953cc8 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":1},{"id":"cb7f960585f1fd1130a88c2e3e6e58e44f24b5ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-02-28 16:22:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3e4ce8224300bc4d4a9f0aecc21689b6d9b27ce1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 18:17:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cf3ff3ab70340e18cfa14b42e4a9d65\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/64abb38bd1ff41e5b45599f11e6f06f6 : SUCCESS in 1h 47m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8f120ecf477412285ede73c74300e8f : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6a91ad8f909456f97a8c73f6db9442c : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/583b73fb9d7243a7b7300d01f7aae8b5 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58bd4a8fe26846bb813a615989eda445 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":2},{"id":"34df29c88b7a5adcfd641a75ed122698e4a6de0c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-04-17 09:40:19.000000000","message":"Abandoned\n\nThis no longer applies to master. We want it stable-only, I guess","accounts_in_message":[],"_revision_number":2},{"id":"b9242698e42288e604b3259f3310c425e0ca21a7","tag":"autogenerated:gerrit:restore","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-04-17 12:25:56.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"2d16af3ee50f8342a9a3bff6b12b8bbe77671aba","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-04-17 12:26:14.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/oslo.db for 875627,2","accounts_in_message":[],"_revision_number":2},{"id":"f6e3904ee5c5e74d15b63e3cf7cc2e9d9558b51e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 12:26:15.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/oslo.db for 875627,2","accounts_in_message":[],"_revision_number":2},{"id":"595f65c2aa62783389a492a989e95fdbdebe1ef3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-04-17 12:28:44.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"f339aed2c37a07d428518d88fd0cc329c635bcb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 14:07:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6edf384870884072860561496350fba7\n\n- oslodb-tox-py39-tips https://zuul.opendev.org/t/openstack/build/45af4252ba7d4d86bdc7c2eea4839137 : FAILURE in 5m 40s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/b3b27f983bd9435ba81d9213bc89d2a9 : FAILURE in 5m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2c299864a7cd44169b0fd5e4b6a3b4c6 : SUCCESS in 1h 31m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07df2beaa6a54bd393712052814b6c60 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11932715e63a4b67b975fa87dfd0ed0f : FAILURE in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/edfad78b1800459f8273110e7c67caaf : FAILURE in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74a902129cde41168069c8382e2dc705 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":3},{"id":"cc2a9a43b7650bac8130ea4df31046e1b9e4aeb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-04-17 14:47:53.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"aed84a76bcda8cfc1e6d2ba6a88ec5e41e91bb52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 15:55:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ba207d5274a4082939eccd9ec983134\n\n- oslodb-tox-py39-tips https://zuul.opendev.org/t/openstack/build/1660b825c3444da7b9f196ecccce6980 : SUCCESS in 5m 09s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/6b044621fb34493796290541ba4e1315 : SUCCESS in 4m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/134b753f0c9c40848def53e137b17764 : SUCCESS in 1h 05m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5534e9c42ed443895330141e0c8f0ed : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a2c13ebcc7b4a4088dfab8c8dd42fbd : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/775fa7ebe20c4667b32ab34e7a66fdb6 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66739165398247bf95a92112e6f3a693 : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":4},{"id":"d6614c23776b4a12da9db54be44d7ae6ff63bee4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-06-02 13:17:37.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"f901be695238b985b2ee7512b2e39c87d4f7ddc3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-06-13 21:03:34.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"7917603f7058364ad080dae6ea9f335bebd5ba86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 21:04:03.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0e984016649b19817db3599508eb8a1ae5a97fba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 22:35:45.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/459c50a78b5648adaf6eef093c4d6f06\n\n- oslodb-tox-py38-tips https://zuul.opendev.org/t/openstack/build/7f3626308cb64de590a07b88fd94219d : FAILURE in 5m 20s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/3ecb9c8bc9f7444498133a7a80f5cad5 : FAILURE in 4m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ec431d995d294bb8a2ba441fe5cdb3ec : SUCCESS in 1h 31m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e73300d63c4f4270b098de1245e74f5f : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e980fcec538400f9c0677d2f53edd32 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d622afe3eb7c4cb59a7ef4e5e19d4de1 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/483f88ea14a44d7ebc1349f56045c110 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34aef0bc63b746e083d32155eab125af : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":4},{"id":"23e0529eb15df30f5c8b0a9a1db7440aa114fd96","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-06-14 23:23:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c912e37054d39a5ee0b1c7b0f041b7b654113a84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 00:39:32.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9eac3b90ee76430d82127c03ceaf2d05\n\n- oslodb-tox-py38-tips https://zuul.opendev.org/t/openstack/build/85221f2818be4155812a747787fb8814 : FAILURE in 4m 58s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/d55a724a898c45b983767e953194b781 : FAILURE in 4m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5e8ba330cd2d42d9ac2247fd0b68c99f : SUCCESS in 1h 15m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8afe487437c84fcaa373e2cd19468f8e : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27a1b7ef448c49df818c4d02e4510cca : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/827a4a8c5d2b4dd98bb7111eecb35499 : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f39957978d34c8daf7519aec33ea55d : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b28eb17995743a48c08485c7494c6fd : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":4},{"id":"6765f9fa8a020aad8733c85a1d17764e7059738f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-06-21 10:51:57.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d2199d48ef0b52b2d37478bde2be90911ae1f820","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 13:22:19.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6671742fcaab4a3d9e2df2c6f9709a27\n\n- oslodb-tox-py38-tips https://zuul.opendev.org/t/openstack/build/352598712ea24b07a7976700c8dc43cc : SUCCESS in 5m 28s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/e861c541e2a54dbaa1190800664ae660 : SUCCESS in 5m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f5351d876fad496d923d35cf068ff7eb : TIMED_OUT in 2h 05m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f27c5ee72464b7c823175f4628f8ccd : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0503d0459ee64c3c861fa2595e004d12 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5671eff96714458f8bcd98db65d1e322 : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4b64eca3b4b4920bf3d397e68b55bcd : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c915526404f0400eb19a0e8e315d54ea : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":4},{"id":"90d7d4ae32e5d0d2fad4335fe708ab6262e12d4a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-06-21 13:27:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2e345bd156b83e3e72dde454b9266583896ffc6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 15:27:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78b7b693a6e141b6947622565ace68e4\n\n- oslodb-tox-py38-tips https://zuul.opendev.org/t/openstack/build/a7a4c032a7c245ea863ee145bf973cdf : SUCCESS in 5m 50s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/9ca939aeeaf24cb99c977cdf963fb0f9 : SUCCESS in 5m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9777b0d6cc4e4b418cc0613e28746490 : SUCCESS in 1h 49m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83a4c1f1235a4311adea6589f76ceef0 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0339c61cdbd44292a1c9dab1901d7b3e : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5cc4441c269341bfbbf3ad1099ad223e : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e646b18757b40beac7989106c7aea00 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6ad53241adb4848ad53e1f608f5a9d0 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":4},{"id":"ba39c71c03a37b8be4be0c8049208877a7033e85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 15:28:10.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"302e18a5b75d2125bdbd24e1dc2db3dd1febf7b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 17:33:50.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/00ede9dfe43147899ba29ec5d7b19978\n\n- oslodb-tox-py38-tips https://zuul.opendev.org/t/openstack/build/23879d35c2ba450c8804c275e8222572 : SUCCESS in 29m 48s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/82b98b71d7294142a7c22013ba03f088 : SUCCESS in 11m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0228a9bdb6ad4e898f847dc3f4c19b42 : TIMED_OUT in 2h 05m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a6ba001a02c43629e8eded9b221c9fb : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0faa4ad7e684892b51aab1efc383650 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c56e215a4a9a48f5883c19d08f8cd2bf : SUCCESS in 5m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dcad2ad3a95e4d058bc32b64b7eae88d : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a23369e05cb4473b5a0a47c25d4bfa6 : SUCCESS in 7m 22s","accounts_in_message":[],"_revision_number":4},{"id":"ef10005f786b0e555830f2fdc55b7d9087033e13","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-06-22 11:59:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b6292b98dd4e4d1ed36f28508fa97118f706bca8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 13:55:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09b5aa2d683741b097e4e80e15098e02\n\n- oslodb-tox-py38-tips https://zuul.opendev.org/t/openstack/build/9a7d11d4280c4024bf74b9a035bd3335 : SUCCESS in 5m 26s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/0699e7ad7d3346d5a30a4d2762133329 : SUCCESS in 5m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bb734671a3714e26bd7c811a82efd3ef : SUCCESS in 1h 43m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72677cfd4247492e907805f31f7f84fc : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64903f03ec534dbbbc50d4ef67a6b3bb : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2942c4862504b6f9101bad59352c36c : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c7e8a73c79ce49ef86aa02b650073430 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc35b6815abe4c94b6e5d52a78a845c9 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":4},{"id":"e1bb9cdb98def4a61327e6999b01209b8bb6445b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 13:56:30.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1addccb70a6cb0b0bee0b8e6d4fb901643d6f1d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 16:13:37.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/222ca0bf4ded472eb3c7f432f08f9936\n\n- oslodb-tox-py38-tips https://zuul.opendev.org/t/openstack/build/112a9a21079d4bf7a68a2136436b39dc : SUCCESS in 27m 19s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/55a4a7a141264decbc83b06ad2236e63 : SUCCESS in 5m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fb7ec31bcfe446bd802a5659f0acd04d : TIMED_OUT in 2h 13m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/863a3785c19b4563aaf852a865602c7c : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88df8f72ec914514a74e556dd3d7342f : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b05672811c444cfbb8206ae122a22ad9 : SUCCESS in 5m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b73e8af1293e4cdc86cbb14da5160d36 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/643d5fff98e348b4af975331c3349620 : SUCCESS in 20m 13s","accounts_in_message":[],"_revision_number":4},{"id":"71363b93ec97afc91004b8290a0be1cd5740fb73","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-06-23 09:45:00.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6766e241d792f1e7ce064fdcbc379e5463b2f437","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 11:23:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c09b243a6934543817777a1326285fe\n\n- oslodb-tox-py38-tips https://zuul.opendev.org/t/openstack/build/1293051df4ce455e8bce32ea4e721396 : SUCCESS in 6m 51s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/9153a715ed1a490a99bc7443b0b85934 : SUCCESS in 6m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2997bb0f28a04bca86efbca4f869a051 : SUCCESS in 1h 36m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1e43d3b34c94ca989f441c7ad5e4dca : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e1d3145c6544b049f65a903865096bc : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8d978b2eaf84e749da0270d5466712e : SUCCESS in 6m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4cc7e5ab35b240bdac56fed4d4096256 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/451941a3a33748ecacbbd534a10ba454 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":4},{"id":"660b02f3b01228e604696cec01e3163a6e2e046f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 11:23:40.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6e056f545281ac318fb683266e952381fd9c7928","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 13:35:45.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30a13cef7ea84367b3c6bf5e3b1a7420\n\n- oslodb-tox-py38-tips https://zuul.opendev.org/t/openstack/build/6ea5b4a5ff0b4c3c907218092d0c9cbb : SUCCESS in 5m 04s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/fa13d8160c0c478d91142fe83cb12f28 : SUCCESS in 5m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8644bf10149948b7af4fbc9868b151bc : SUCCESS in 2h 03m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c37f7bb4bf49451ca87ada1a7597aff6 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06feffb3abbd4fbf817c6e797b3b4f62 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b11c266dec3645cbbd2daa6bff5aec24 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d9e6863c16754698b40a4f4465279f24 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20b143e2d221404a9f4d7cd417727ae1 : SUCCESS in 13m 03s","accounts_in_message":[],"_revision_number":4},{"id":"dc4c7b80e850dd0245abf64551498efcc011a9b3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 13:35:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"135e7db8a39d7239eb010883aaac038746cd5f6d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 13:36:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a903eb55f7e244d58d46e3652f16d45c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/038f115e16784abe82f19bc451615a0d : SUCCESS in 43s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f6ba070657b3bb6b57cef3728b3e6b95f0bf08cb","revisions":{"b1dfb5333d02a74b0bafad7a2ed444c7494d96ab":{"kind":"REWORK","_number":1,"created":"2023-02-28 10:21:23.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/27/875627/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/27/875627/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/27/875627/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/27/875627/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/27/875627/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/27/875627/1"}}},"commit":{"parents":[{"commit":"5cd7962a2d35902703aac3b68b85e50e26ff4981","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/5cd7962a2d35902703aac3b68b85e50e26ff4981"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-02-27 17:23:31.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-02-27 17:30:38.000000000","tz":0},"subject":"Ignore unset opts when deprecated","message":"Ignore unset opts when deprecated\n\nIf an option is deprecated, there\u0027s a good chance we want to ignore it\nunless explicitly requested by the user. Start ignoring options that are\ndeprecated and unset.\n\nChange-Id: I3008562db52a5625504b262d7148528a832246dd\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nCloses-Bug: #2008720\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/b1dfb5333d02a74b0bafad7a2ed444c7494d96ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/b1dfb5333d02a74b0bafad7a2ed444c7494d96ab"}]},"branch":"refs/heads/master"},"8a03e8207d8edb15f19d16cdec9f696194b3c640":{"kind":"REWORK","_number":2,"created":"2023-02-28 16:22:32.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/27/875627/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/27/875627/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/27/875627/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/27/875627/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/27/875627/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/27/875627/2"}}},"commit":{"parents":[{"commit":"5cd7962a2d35902703aac3b68b85e50e26ff4981","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/5cd7962a2d35902703aac3b68b85e50e26ff4981"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-02-27 17:23:31.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-02-28 16:20:46.000000000","tz":0},"subject":"Ignore unset opts when deprecated","message":"Ignore unset opts when deprecated\n\nIf an option is deprecated, there\u0027s a good chance we want to ignore it\nunless explicitly requested by the user. Start ignoring options that are\ndeprecated and unset.\n\nChange-Id: I3008562db52a5625504b262d7148528a832246dd\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nCloses-Bug: #2008720\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/8a03e8207d8edb15f19d16cdec9f696194b3c640"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/8a03e8207d8edb15f19d16cdec9f696194b3c640"}]},"branch":"refs/heads/master"},"5afe56507861c7fe7d1c14a736d8ed34697c3878":{"kind":"REWORK","_number":3,"created":"2023-04-17 12:28:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/27/875627/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/27/875627/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/27/875627/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/27/875627/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/27/875627/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/27/875627/3"}}},"commit":{"parents":[{"commit":"c3305cb282de8a7686a648a24a9fe3b6cd5d999c","subject":"Merge \"Remove legacy base test classes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/c3305cb282de8a7686a648a24a9fe3b6cd5d999c"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-02-27 17:23:31.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-04-17 12:28:36.000000000","tz":60},"subject":"Ignore unset opts when deprecated","message":"Ignore unset opts when deprecated\n\nIf an option is deprecated, there\u0027s a good chance we want to ignore it\nunless explicitly requested by the user. Start ignoring options that are\ndeprecated and unset.\n\nChange-Id: I3008562db52a5625504b262d7148528a832246dd\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nCloses-Bug: #2008720\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/5afe56507861c7fe7d1c14a736d8ed34697c3878"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/5afe56507861c7fe7d1c14a736d8ed34697c3878"}]},"branch":"refs/heads/master"},"f6ba070657b3bb6b57cef3728b3e6b95f0bf08cb":{"kind":"REWORK","_number":4,"created":"2023-04-17 14:47:53.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/27/875627/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/27/875627/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/27/875627/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/27/875627/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/27/875627/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/27/875627/4"}}},"commit":{"parents":[{"commit":"c3305cb282de8a7686a648a24a9fe3b6cd5d999c","subject":"Merge \"Remove legacy base test classes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/c3305cb282de8a7686a648a24a9fe3b6cd5d999c"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-02-27 17:23:31.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-04-17 14:47:53.000000000","tz":0},"subject":"Ignore unset opts when deprecated","message":"Ignore unset opts when deprecated\n\nIf an option is deprecated, there\u0027s a good chance we want to ignore it\nunless explicitly requested by the user. Start ignoring options that are\ndeprecated and unset.\n\nChange-Id: I3008562db52a5625504b262d7148528a832246dd\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nCloses-Bug: #2008720\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f6ba070657b3bb6b57cef3728b3e6b95f0bf08cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f6ba070657b3bb6b57cef3728b3e6b95f0bf08cb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
