)]}'
{"id":"openstack%2Foslo.db~878213","triplet_id":"openstack%2Foslo.db~master~I4e1faa2c617ac19e7c9766e99ee9012ad9298d31","project":"openstack/oslo.db","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-04-17 12:31:59.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I4e1faa2c617ac19e7c9766e99ee9012ad9298d31","subject":"Remove dead code","status":"MERGED","created":"2023-03-22 12:26:09.000000000","updated":"2023-04-17 14:04:54.000000000","submitted":"2023-04-17 14:04:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878213-sqlalchemy-20","meta_rev_id":"9dedd419f1d337e6320f6776a47703e8d8f20003","_number":878213,"virtual_id_number":878213,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-17 14:04:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2023-04-17 11:53:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2023-04-17 12:31:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-22 12:26:24.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-03-22 14:05:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-17 11:53:37.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"747a1c78a73f9e899687930ba4301cc4606e3149","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-03-22 12:26:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ebc42e41ee0d819dd8f7cc04966c338d3c46d25","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-22 12:26:24.000000000","message":"Patch Set 1:\n\nBuild ${ISCSI_JOB} ${env.ISCSI_JOB}/${env.ISCSI_JOB_NUMBER} finished with result ${env.ISCSI_JOB_RESULT}. Logs link: ${env.ISCSI_URL}","accounts_in_message":[],"_revision_number":1},{"id":"aef22dafdf60a7369f3d49718b3c5efc218d50dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 14:05:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6f6696e1b694133b0daeaeb448cd694\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a23f06a349a5422f835d0fd3ea38ad08 : SUCCESS in 1h 32m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a442af30f6b40d38233b6b7d4236811 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00460a709c4f4aa2a7751d637aef4565 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2741b11ae43042c8a6b33bea3e40b38b : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/854e8079b4f544eea1578396f662b03d : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":1},{"id":"b3e6106a999935aad750940ddfd252a9306fb2c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-04-05 12:50:41.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"501a769a499e3497179119b940c1d3f3960e8ca1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 14:26:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aea34413c22145eda50b06922cb64fed\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ef3fcdc5db264f9da200543a5eb1f400 : SUCCESS in 1h 29m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e88ef90d500c4cb1b72eb6ae738c0c1a : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95fcbd2b7a614764865eb814fa5f46d6 : FAILURE in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4a05dcbeae9343d983d39953c4efa77c : FAILURE in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24407a27402242e4b0f06f8fdeb61994 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":2},{"id":"ec800c03a8a5182e9f2b931856d50005b3e45965","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-04-17 09:19:18.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"79b7f8508383e8ec39781750a0d95a2348ac4499","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 10:22:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a001789075e44485a5fb49870b52a632\n\n- oslodb-tox-py39-tips https://zuul.opendev.org/t/openstack/build/6bf3dc2db069493f916c5a276e743359 : SUCCESS in 5m 31s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/aa8ee42a42004d3eb15fa91109b86569 : SUCCESS in 4m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/666d39191ad44f1caecc3b6d89023701 : SUCCESS in 59m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7727a4cece6c47cd99cf90bdb04359a7 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f00f51e532964fd282eebdfd6babf552 : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/37e0e71673ec41f6ac0a3c7df0623032 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f2bd749afe24d789102b67e41b45480 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":3},{"id":"9a3735774f8b4a0f4c06edb5b3c9ace17a1cb0da","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-04-17 11:53:37.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"4887b057cc7e076a8a4743a74311c3a4be9f9578","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-04-17 12:31:59.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4b6187554184da0f118664ea206adfd63a08a8f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 12:32:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ff130452601fea16d71fa798b2cba255deb5da81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 14:04:54.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9dd7b77294e4450ab05e9c11628ce58\n\n- oslodb-tox-py39-tips https://zuul.opendev.org/t/openstack/build/b3e7af1235694243a054f5f29a1eeb8f : SUCCESS in 5m 18s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/ad817949a01f401ca16efb6ad1e8be28 : SUCCESS in 4m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7651d79b03d241a0a430539684f25816 : SUCCESS in 1h 22m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0a8b83cd0b74e149834a6c923d63287 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc90e39d0add4e698a1bf6fc40a3da82 : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/82fe3d6793a349af9ed10f2ea3f6edb4 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8476378a10d9467c83930fcfae76374a : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":3},{"id":"9dedd419f1d337e6320f6776a47703e8d8f20003","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 14:04:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"04acea8cbbf7a989e07dad66fd40ac1d3ba4e0d2","revisions":{"ea051f4d5a8dd7594d5b6d4a0707c0b798b811e8":{"kind":"REWORK","_number":1,"created":"2023-03-22 12:26:09.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/13/878213/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/13/878213/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/13/878213/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/13/878213/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/13/878213/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/13/878213/1"}}},"commit":{"parents":[{"commit":"94d6e24ca19b0116eed00d5ccdb8a538918c6dcf","subject":"Remove sqlalchemy-migrate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/94d6e24ca19b0116eed00d5ccdb8a538918c6dcf"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-03-22 12:04:29.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-03-22 12:04:29.000000000","tz":0},"subject":"Remove dead code","message":"Remove dead code\n\nThis should have been removed in change\nIc3d6bd318038d723b0d50d39e45f8e26289e9a57 but was missed.\n\nChange-Id: I4e1faa2c617ac19e7c9766e99ee9012ad9298d31\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/ea051f4d5a8dd7594d5b6d4a0707c0b798b811e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/ea051f4d5a8dd7594d5b6d4a0707c0b798b811e8"}]},"branch":"refs/heads/master"},"9942934185650664441ea848cdcf6b564c278254":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-04-05 12:50:41.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/13/878213/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/13/878213/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/13/878213/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/13/878213/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/13/878213/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/13/878213/2"}}},"commit":{"parents":[{"commit":"69c90d8cee615be987ef1c949ee12ae39eaa4407","subject":"Don\u0027t sleep in tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/69c90d8cee615be987ef1c949ee12ae39eaa4407"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-03-22 12:04:29.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-04-05 12:48:43.000000000","tz":60},"subject":"Remove dead code","message":"Remove dead code\n\nThis should have been removed in change\nIc3d6bd318038d723b0d50d39e45f8e26289e9a57 but was missed.\n\nChange-Id: I4e1faa2c617ac19e7c9766e99ee9012ad9298d31\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/9942934185650664441ea848cdcf6b564c278254"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/9942934185650664441ea848cdcf6b564c278254"}]},"branch":"refs/heads/master"},"04acea8cbbf7a989e07dad66fd40ac1d3ba4e0d2":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-04-17 09:19:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/13/878213/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/13/878213/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/13/878213/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/13/878213/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/13/878213/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/13/878213/3"}}},"commit":{"parents":[{"commit":"672dd056aa66645648c0d71af5eae3da9f424e48","subject":"Don\u0027t sleep in tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/672dd056aa66645648c0d71af5eae3da9f424e48"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-03-22 12:04:29.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-04-17 09:17:38.000000000","tz":60},"subject":"Remove dead code","message":"Remove dead code\n\nThis should have been removed in change\nIc3d6bd318038d723b0d50d39e45f8e26289e9a57 but was missed.\n\nChange-Id: I4e1faa2c617ac19e7c9766e99ee9012ad9298d31\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/04acea8cbbf7a989e07dad66fd40ac1d3ba4e0d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/04acea8cbbf7a989e07dad66fd40ac1d3ba4e0d2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
