)]}'
{"id":"openstack%2Foslo.db~878216","triplet_id":"openstack%2Foslo.db~master~Ice3a0c0d0a5b8c2920c7f775ff8ce974b572c66e","project":"openstack/oslo.db","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-04-17 12:32:02.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"Ice3a0c0d0a5b8c2920c7f775ff8ce974b572c66e","subject":"Don\u0027t sleep in tests","status":"MERGED","created":"2023-03-22 12:26:09.000000000","updated":"2023-04-17 14:06:06.000000000","submitted":"2023-04-17 14:04:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878216-sqlalchemy-20","meta_rev_id":"0d857d22798ccb911773e9c50805d6cec9eb0c79","_number":878216,"virtual_id_number":878216,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-17 14:04:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2023-04-17 11:52:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2023-04-17 12:32:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-22 12:26:25.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2023-03-22 13:46:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-17 11:52:33.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"8e28c1ff758dda78c19f34d807734673045be5fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-03-22 12:26:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0491f3befd92e8a8980c1d4223229e97b33a6671","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-22 12:26:25.000000000","message":"Patch Set 1:\n\nBuild ${ISCSI_JOB} ${env.ISCSI_JOB}/${env.ISCSI_JOB_NUMBER} finished with result ${env.ISCSI_JOB_RESULT}. Logs link: ${env.ISCSI_URL}","accounts_in_message":[],"_revision_number":1},{"id":"4d35e945f941b8be4e33f1a7c61fcd9c4320fc3f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2023-03-22 12:32:10.000000000","message":"Patch Set 1:\n\nBuild ${ISCSI_JOB} ${env.ISCSI_JOB}/${env.ISCSI_JOB_NUMBER} finished with result ${env.ISCSI_JOB_RESULT}. Logs link: ${env.ISCSI_URL}","accounts_in_message":[],"_revision_number":1},{"id":"04a8810c4da51ceaf14077a030ca488bf848ab94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 13:46:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6c32899eb5b041a4b439155cc379a72b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/689ce852da7549398cd6ad2935d6554a : SUCCESS in 1h 17m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cad2ddbd045749138b17e9d09cd8c943 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/981b653332c242908316d39c395ffae4 : FAILURE in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a3ab1e8ae4d04d41b065e08e050c18ea : FAILURE in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14ed763e9c194fca8e5b8f387eb5916f : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":1},{"id":"51d3657a497d8c5bb4c7739907495954aad83536","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-04-05 12:50:41.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"6c4e120e29322fa10b27f93aa9bfcaf5c2398039","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 14:16:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0583981c9c5347dbafbcd2ce73571d19\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9c36585e0038424e884a2417ef9654b7 : SUCCESS in 1h 20m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b7abb8d5bcc49b4be29f84d494c6d5b : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72db649d634440b7b59c5c1f86dd9dae : FAILURE in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02e01b6ecf81480d9e777dc30506bd63 : FAILURE in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/278bf9af6faa4638ad5bbeb03b7b33f5 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":2},{"id":"52892e45ae604383b2989f719b8c3cee4cc4a2db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-04-17 09:19:18.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"2fe1d846a951a04188cea90fcc97db6640ce1b88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 11:09:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00420780079f48f1862b5d1413cc57ee\n\n- oslodb-tox-py39-tips https://zuul.opendev.org/t/openstack/build/ed2248181faf465aa0b48a2f77c5603d : SUCCESS in 5m 52s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/3cfcd860042645b6adbe4a411a9d18e6 : SUCCESS in 4m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1a3046292a6e45a0aeb94c8125ec3845 : SUCCESS in 1h 42m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90167a31dcbf4074a309926f63fb8da6 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c005644f6f224856b6057bd951838039 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/22a896ad7301427aac84201af6a64289 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcb9869407914d21bac92c3588fbea0f : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":3},{"id":"117f269075a4f945fb01395151c122c2d50815bc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-04-17 11:52:33.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ae20d5e0b451d68ee3797bb1eb364b636a0e849f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-04-17 12:32:02.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d202a325dedba88245c802468d64199826c8a415","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 12:32:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"54a3ba5a6259a7e6b2455de26297687c88b582f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 14:04:52.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/400de6527a86483ba4810db9afc559c4\n\n- oslodb-tox-py39-tips https://zuul.opendev.org/t/openstack/build/ab9d862696fe4b0ba3838467ef497db7 : SUCCESS in 5m 16s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/d292ba6597b149f6bd9773c91ee5b882 : SUCCESS in 5m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fd334ac896d5476ca9ff653198505076 : SUCCESS in 1h 27m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb9fb9b0d68f4a42b3ac8afecd2d9433 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/597d390131104feb807379a79fd22b56 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/80b3cbc0eda64eaaaf497f287b87d200 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7c0a4fb550f486284177b26b965c397 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":3},{"id":"3db438b300e1854b6865f999a562c32b7205f78c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 14:04:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"0d857d22798ccb911773e9c50805d6cec9eb0c79","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 14:06:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50cfcbdbcc734167a7f1b2c98b572059\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7955e34cea3a4677820d21e87204ed81 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"672dd056aa66645648c0d71af5eae3da9f424e48","revisions":{"05a4a065be6e335580972be1ea3584907656f1f8":{"kind":"REWORK","_number":1,"created":"2023-03-22 12:26:09.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/16/878216/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/16/878216/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/16/878216/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/16/878216/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/16/878216/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/16/878216/1"}}},"commit":{"parents":[{"commit":"f058d86ee96f59690be5133a8f6d5668f25cf62e","subject":"Remove logic for SQLAlchemy \u003c 1.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f058d86ee96f59690be5133a8f6d5668f25cf62e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-03-22 12:22:03.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-03-22 12:25:49.000000000","tz":0},"subject":"Don\u0027t sleep in tests","message":"Don\u0027t sleep in tests\n\nMock it out, reducing test run of ~3 and ~15 seconds to milliseconds.\n\nChange-Id: Ice3a0c0d0a5b8c2920c7f775ff8ce974b572c66e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/05a4a065be6e335580972be1ea3584907656f1f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/05a4a065be6e335580972be1ea3584907656f1f8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f058d86ee96f59690be5133a8f6d5668f25cf62e","is_merged_in_target_branch":false,"change_id":"I4d4a58e15e840ecfa63e15c709617a65642c8323","change_number":878215,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"69c90d8cee615be987ef1c949ee12ae39eaa4407":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-04-05 12:50:41.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/16/878216/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/16/878216/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/16/878216/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/16/878216/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/16/878216/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/16/878216/2"}}},"commit":{"parents":[{"commit":"da4f13e7345653eba8aab5b8aceeaeff7367989e","subject":"Do not convert to string ``url.URL`` objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/da4f13e7345653eba8aab5b8aceeaeff7367989e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-03-22 12:22:03.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-04-05 12:48:43.000000000","tz":60},"subject":"Don\u0027t sleep in tests","message":"Don\u0027t sleep in tests\n\nMock it out, reducing test run of ~3 and ~15 seconds to milliseconds.\n\nChange-Id: Ice3a0c0d0a5b8c2920c7f775ff8ce974b572c66e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/69c90d8cee615be987ef1c949ee12ae39eaa4407"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/69c90d8cee615be987ef1c949ee12ae39eaa4407"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"da4f13e7345653eba8aab5b8aceeaeff7367989e","is_merged_in_target_branch":true,"change_id":"I84f13f378f83e2a55078370ae2b4787f00982c23","change_number":878642,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"672dd056aa66645648c0d71af5eae3da9f424e48":{"kind":"REWORK","_number":3,"created":"2023-04-17 09:19:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/16/878216/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/16/878216/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/16/878216/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/16/878216/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/16/878216/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/16/878216/3"}}},"commit":{"parents":[{"commit":"481936a822425c6a8c806409c8c877eed6661a2c","subject":"Moves supported python runtimes from version 3.8 to 3.10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/481936a822425c6a8c806409c8c877eed6661a2c"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-03-22 12:22:03.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-04-17 09:17:35.000000000","tz":60},"subject":"Don\u0027t sleep in tests","message":"Don\u0027t sleep in tests\n\nMock it out, reducing test run of ~3 and ~15 seconds to milliseconds.\n\nChange-Id: Ice3a0c0d0a5b8c2920c7f775ff8ce974b572c66e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/672dd056aa66645648c0d71af5eae3da9f424e48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/672dd056aa66645648c0d71af5eae3da9f424e48"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"481936a822425c6a8c806409c8c877eed6661a2c","is_merged_in_target_branch":true,"change_id":"I0cb0d4e03858a4f394aed484508b305683ce7863","change_number":879281,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
