)]}'
{"id":"openstack%2Foslo.db~888580","triplet_id":"openstack%2Foslo.db~master~Ibd3f665a3ed3aedf9d1f33edcab35a46c27ea3dc","project":"openstack/oslo.db","branch":"master","topic":"exception-issues","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-07-18 17:29:34.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"11816":{"account":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"last_update":"2023-07-18 13:44:08.000000000","reason":"\u003cGERRIT_ACCOUNT_11816\u003e replied on the change","reason_account":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}}},"hashtags":[],"change_id":"Ibd3f665a3ed3aedf9d1f33edcab35a46c27ea3dc","subject":"exc_filters: Do not reraise","status":"MERGED","created":"2023-07-14 14:41:50.000000000","updated":"2023-07-19 13:47:00.000000000","submitted":"2023-07-19 13:46:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888580-exception-issues","meta_rev_id":"6641d967c4fec87762c0371814b713536284bc88","_number":888580,"virtual_id_number":888580,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-19 13:45:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2023-07-18 13:44:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2023-07-19 11:49:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-14 14:44:17.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2023-07-14 16:53:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"36346661e6cb59287f10adbadc80fe6cf718ae4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-14 14:41:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36e4efd1e275053b996604fd48def290dc73489b","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2023-07-14 16:22:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5de9c9eb921043c0e9a580a72fa38ec6f0602bb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-14 16:53:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/88ef8cb84eec41efa9434f53f078b8af\n\n- oslodb-tox-py38-tips https://zuul.opendev.org/t/openstack/build/fedcaa8b62e04c40a90cfb5fa6b1039f : SUCCESS in 5m 35s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/2279c8d7902944199623534825aab449 : SUCCESS in 5m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b0bf6cc4b7724dbb92d83ff6d171b990 : TIMED_OUT in 2h 04m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fb7d26c8d09470eb41a8b54af41748a : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1589f3fade1c40cbbc22a3ba84bdf5a6 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49e56fef70174bdf82d858e824d26aae : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6bbd97ac11c14e3185b698b4f040fbc6 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0529da5d55f64569b8190b7412c21120 : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":1},{"id":"34436192136bdff7633eebd434bddb2f68064786","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-17 09:34:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4fa290e93ff78b04469ed0f5981fa8f10f375d38","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-17 09:38:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f67aa0f89e8ff4c893da7b9d3ae6ae494937044","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-17 09:39:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3384c3adf0fdc03e7bead25f1de66907ea20ad7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 11:52:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ec7d0952435456dbb5781efbdd8ce10\n\n- oslodb-tox-py38-tips https://zuul.opendev.org/t/openstack/build/48553de79057473090d4c950eb8576f8 : SUCCESS in 5m 51s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/1570ee81f95c4a9da0d5729a88c31fa6 : SUCCESS in 5m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4b946b514c7e409f8dcd640e9741d218 : TIMED_OUT in 2h 04m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9004570b53254f08a38d8b3ad141cd4a : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2220b638f46c4b0cbcecec7f6adc50a8 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/034e553beac64eebb20f463135892fd6 : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a82a1673a4a8495c8b56862e452482c4 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9467a5264294516b42a3787e22269ee : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":1},{"id":"a87477b949f942b312e5dabf5b844dd26735da83","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2023-07-17 15:03:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c51b242c3c56f935acadf2c5e2162a6d58f52c60","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-18 11:03:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f678ad93f18f8a30c0c7c9302401c295065d2b4","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2023-07-18 13:44:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f7ecfe780e0c79e132b43a5daf442abcd9daee5b","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2023-07-18 13:44:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"209717e3bf29019a99e7737153c98e3e9a9bc456","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-18 17:29:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"95a2c0c9d8b479e8b1f6573d324f75e2aeb3b3e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-18 19:19:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9186dfdcb9614d9aa426dfcf08e7d260\n\n- oslodb-tox-py38-tips https://zuul.opendev.org/t/openstack/build/a063505e915f44f8bddb83ed239123ab : SUCCESS in 6m 50s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/060814c44e9b4111a8660f57cbdd10d8 : SUCCESS in 5m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bab9c5fbcabf41fba022990d1becfc18 : SUCCESS in 1h 44m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbd7edb5155a4643a81ff3c48c612a79 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f309726468b847e8926310649019fc6d : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e98a0d5d18c4401997689de023c34ad8 : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/019aa4f7db794c7196b3c4ca64207ac4 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03493cf97b4f4579b99c0731dea1d280 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"974e98a3ff2799728fc063628512199593fc8948","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-19 11:49:48.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"47f1d6267fb34e02d4a4aaa646dc564029c7463a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 11:50:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d7224ec539331f20439100da9ab383cae2d7c0d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 13:45:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87acf39b72bf4b9aa5c9f904662c3511\n\n- oslodb-tox-py38-tips https://zuul.opendev.org/t/openstack/build/0672b529e0f443ca9723bdfdd1bd7401 : SUCCESS in 6m 58s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/16eaf83c223a44e5b4aafcbe43e21fc1 : SUCCESS in 6m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2baa4f8e222748b6baf13e13bd9c54a6 : SUCCESS in 1h 48m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5be4a82bae694cb9812e5bc0ca81a6c4 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d725477369404976a3beb36b368caf0a : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bdec956d935b4c909c7e598f54f4f4ae : SUCCESS in 5m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/456362179cc248f3846f1ba590b31386 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/380978badd1d4ba8904e5969db057a5f : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":1},{"id":"92fcb9be2a67912de93445d4bf6ac2863e26ef48","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 13:46:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6641d967c4fec87762c0371814b713536284bc88","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 13:47:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb6270a802584c818077ae203291e4fb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ade857e60c844a7ae6727f5fc3c2316 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8dcd47383d35a0f4e2283b09d70f34ad120a85f7","revisions":{"8dcd47383d35a0f4e2283b09d70f34ad120a85f7":{"kind":"REWORK","_number":1,"created":"2023-07-14 14:41:50.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/80/888580/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/80/888580/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/80/888580/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/80/888580/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/80/888580/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/80/888580/1"}}},"commit":{"parents":[{"commit":"6285f2b4e2ec9129b898ab2d5ad2117bc8ff3c73","subject":"Merge \"Ignore unset opts when deprecated\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/6285f2b4e2ec9129b898ab2d5ad2117bc8ff3c73"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-07-14 14:39:47.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-07-14 14:41:37.000000000","tz":60},"subject":"exc_filters: Do not reraise","message":"exc_filters: Do not reraise\n\nWe\u0027re seeing a weird condition whereby the exceptions raised by our\nexception handler and being \"rehandled\", resulting in \u0027DBError\u0027 being\nraised instead of e.g. \u0027DBReferenceError\u0027. This is hard to reproduce and\nthus far has only be spotted in Barbican unit tests. Prevent the issue\nby checking if the exception being handled is one of our own and skip\nhandling it if so.\n\nChange-Id: Ibd3f665a3ed3aedf9d1f33edcab35a46c27ea3dc\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/8dcd47383d35a0f4e2283b09d70f34ad120a85f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/8dcd47383d35a0f4e2283b09d70f34ad120a85f7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
