)]}'
{"id":"openstack%2Foslo.db~889081","triplet_id":"openstack%2Foslo.db~master~Id59e3fa488a982943415a04aeeefb8e6aaddf467","project":"openstack/oslo.db","branch":"master","topic":"echo-parameter","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-07-26 15:53:48.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"11816":{"account":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"last_update":"2023-07-26 15:53:50.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id59e3fa488a982943415a04aeeefb8e6aaddf467","subject":"WIP: Add ability to configure autobegin attribute","status":"ABANDONED","created":"2023-07-20 10:52:12.000000000","updated":"2023-07-26 15:53:50.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4ea696f0df665357ee3bd6776d5da69fda2d51ba","_number":889081,"virtual_id_number":889081,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-07-20 11:26:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-20 11:26:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-20 13:43:06.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"CC"}],"messages":[{"id":"d1e56f4b4c2171ce6bab3a0639a5c96b8d10637d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-20 10:52:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b1764ca733923a99c76720fa3853a945d9137e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 11:26:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a92c29d296ce4e1fbcca5a4b7a7fea7a\n\n- oslodb-tox-py38-tips https://zuul.opendev.org/t/openstack/build/68ff52077d5d4320922ea17ddd6cf2a3 : FAILURE in 5m 10s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/7157257ae11f434c9076049a0d8b6413 : FAILURE in 3m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ef5447611a6642c683c113421e4d5d6b : SUCCESS in 3m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2757055bfbf1405bb1647f69b1e66a7a : FAILURE in 24m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/888656ad79474e9e9191a06d46b1718d : FAILURE in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec069657d37145d1bb39e5e772eec400 : FAILURE in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b80ba1f0a4a48d1b55ddf2cafe46eca : FAILURE in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d8e435ba114a4a2f9f15d9f3e2fc0ff1 : FAILURE in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10ee65e945354927bf2e0125c371b2ed : FAILURE in 3m 49s","accounts_in_message":[],"_revision_number":1},{"id":"878fe973f8f96310853a072466db7ee869b23baa","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2023-07-20 13:43:06.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"07c330cc40a5f936278288b9278bf119f2f46dbd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-26 15:53:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4ea696f0df665357ee3bd6776d5da69fda2d51ba","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-26 15:53:50.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"66d8d46206e649bc33844d15b25f7a6e05c52c9a","revisions":{"66d8d46206e649bc33844d15b25f7a6e05c52c9a":{"kind":"REWORK","_number":1,"created":"2023-07-20 10:52:12.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/81/889081/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/81/889081/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/81/889081/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/81/889081/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/81/889081/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/81/889081/1"}}},"commit":{"parents":[{"commit":"008f09bc0e44b0ba69409b01a362f1038cc199e6","subject":"Add ability to configure \u0027echo\u0027 attribute","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/008f09bc0e44b0ba69409b01a362f1038cc199e6"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-07-14 12:55:20.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-07-20 10:51:35.000000000","tz":60},"subject":"WIP: Add ability to configure autobegin attribute","message":"WIP: Add ability to configure autobegin attribute\n\nThis is helpful while migrating to SQLAlchemy 2.x. While we\u0027re here, an\nunnecessary \u0027autocommit\u0027 parameter is removed: this must always be false\nin SQLAlchemy 2.x, which we support [1].\n\n[1] https://docs.sqlalchemy.org/en/20/orm/session_api.html#sqlalchemy.orm.Session.params.autocommit\n\nWIP because this is only supported in 2.0 and that\u0027s blocked by\nupper-constraints right now.\n\nChange-Id: Id59e3fa488a982943415a04aeeefb8e6aaddf467\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/66d8d46206e649bc33844d15b25f7a6e05c52c9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/66d8d46206e649bc33844d15b25f7a6e05c52c9a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
