)]}'
{"id":"openstack%2Foslo.db~894207","triplet_id":"openstack%2Foslo.db~master~Id5f64a5cedd45769188be12fb5df8e43b2abcb1d","project":"openstack/oslo.db","branch":"master","topic":"py311","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-09-14 03:29:11.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-09-14 03:29:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id5f64a5cedd45769188be12fb5df8e43b2abcb1d","subject":"bindep: Use new mysql-* package names","status":"MERGED","created":"2023-09-07 16:23:36.000000000","updated":"2023-09-14 08:35:10.000000000","submitted":"2023-09-14 03:29:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894207-py311","meta_rev_id":"de083dbb34d1af3ea53ad64f3bfd38ef406a4876","_number":894207,"virtual_id_number":894207,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-14 03:29:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2023-09-14 01:48:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2023-09-14 08:35:10.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2023-09-14 01:48:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-07 16:48:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-11 09:02:16.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-09-14 01:48:25.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"9ca6a121fda25e6f7002371ed280f160897647a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-07 16:23:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92cfcbb684eebc62818ca715918e90ca12d3a86c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-07 16:23:57.000000000","message":"Topic python-311 removed","accounts_in_message":[],"_revision_number":1},{"id":"2b53f5d4bd90b0563fef2975e01b98547273fe26","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-07 16:24:02.000000000","message":"Topic set to py311","accounts_in_message":[],"_revision_number":1},{"id":"df67690128dbb76f82aebe6b18e366474f96a81f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 16:48:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8010355e30164e36b53351c13fb508d5\n\n- oslodb-tox-py38-tips https://zuul.opendev.org/t/openstack/build/a1de109bf69f4f12b16446475f7d85f7 : RETRY_LIMIT in 2m 17s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/e9df022cfe614c9f9b5be000fc32ede2 : RETRY_LIMIT in 2m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0fe607314106447192975ffd611069a2 : FAILURE in 10m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06068676ac6e4d9f91f9a03c4835ca0a : RETRY_LIMIT in 2m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76cef970c6d546ba93f05d44485ba76b : RETRY_LIMIT in 1m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a8790e309f348a69526a60f17971be7 : RETRY_LIMIT in 1m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e2de21454794e97afa891182c9dbe12 : RETRY_LIMIT in 2m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/31bedd6dd10d4d1c8a93f52992c23ae1 : RETRY_LIMIT in 3m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cfca01c055149698d5ae6de404cf4fd : RETRY_LIMIT in 1m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/972369ff13974ee78ccf5effbc2f8d1c : RETRY_LIMIT in 1m 46s","accounts_in_message":[],"_revision_number":1},{"id":"fbadf843480afc17fa31f44240a68006ee878766","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-08 11:19:04.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f224b4b10216dc9d94b4832195cc58f8e6619013","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 13:08:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff9acedb9c024af28bb6adaf017d4419\n\n- oslodb-tox-py38-tips https://zuul.opendev.org/t/openstack/build/c34748480c5e43deb02c9ad1c455d9fd : SUCCESS in 5m 34s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/8fc86a0666634d6f8cdafe955245154b : SUCCESS in 5m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/715b7adae6414eef98334decbc9939b2 : SUCCESS in 1h 41m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/740e3190af9743039c1764f8dd3b0ec2 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/906ea46f3fdd4edda3771bb39e4b5cac : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e624dd813c1f435f855143a16f5fe640 : SUCCESS in 5m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/95a0b0306db84410a2f0f506a22d3e01 : SUCCESS in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8d2c7281f4174765ad993417e5ddb24f : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f230f40c4cab4bd8b3853538196ae715 : SUCCESS in 4m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c871aa75559d4824bfbe39cb55d3ac5f : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":2},{"id":"92bd2f497f1620c6376f5d16f54fe2b596b84ece","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-14 01:48:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ca651b4d93ebc45a5361007a2fb0629d0317e0cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 01:48:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5dcf940f9181e4194b4cd6639a0397390d43f7d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 03:29:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b0b7baad1774cef9d9654e0a3224a3c\n\n- oslodb-tox-py38-tips https://zuul.opendev.org/t/openstack/build/a725a3e5d34645a79c4389767d111190 : SUCCESS in 5m 50s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/2df94f20168f424280f467dbfe8f0877 : SUCCESS in 9m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f683e9a5e18f4beab2772720e9064218 : SUCCESS in 1h 33m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca18d52bbdda478b8104b8433dbc290d : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70b5568048c94bec93856e1b7c8e88a8 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8006acf63ba54e2d80ae52adb5be67c4 : SUCCESS in 5m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/356c11ab44f14dbab8d251ad25dd8324 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/321c4c8a7db141fb8f4d7d9b0880292e : SUCCESS in 3m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9fe3073ebf54cd1a6ffade05bc72126 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":2},{"id":"8a912a9e5916fcbbc3c57f097f557b8733df0b49","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 03:29:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"481a454668e01a12d0b90bd01aadc4519a7c0050","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 03:30:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4dbf8b87b4d344cab813c94b449d8d97\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63014422da8b45e3870e200fa427e9aa : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3969761cb1d74b5abbe534593eec3746 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2},{"id":"de083dbb34d1af3ea53ad64f3bfd38ef406a4876","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-09-14 08:35:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5854e6b0bb3ab47ace8467c2c6da659c5f31af07","revisions":{"e68ec8cd4b17f61b3a0c83819382e44f3fa703d2":{"kind":"REWORK","_number":1,"created":"2023-09-07 16:23:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/07/894207/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/07/894207/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/07/894207/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/07/894207/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/07/894207/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/07/894207/1"}}},"commit":{"parents":[{"commit":"1ae346824c3583f5ab701e3d0bc073542c3ed91a","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/1ae346824c3583f5ab701e3d0bc073542c3ed91a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-07 16:22:20.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-07 16:22:20.000000000","tz":60},"subject":"bindep: Use new mysql-* package names","message":"bindep: Use new mysql-* package names\n\nDebian Bookworm uses a different package name for the server and client\npackages.\n\nChange-Id: Id5f64a5cedd45769188be12fb5df8e43b2abcb1d\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e68ec8cd4b17f61b3a0c83819382e44f3fa703d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e68ec8cd4b17f61b3a0c83819382e44f3fa703d2"}]},"branch":"refs/heads/master"},"5854e6b0bb3ab47ace8467c2c6da659c5f31af07":{"kind":"REWORK","_number":2,"created":"2023-09-08 11:19:04.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/07/894207/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/07/894207/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/07/894207/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/07/894207/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/07/894207/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/07/894207/2"}}},"commit":{"parents":[{"commit":"1ae346824c3583f5ab701e3d0bc073542c3ed91a","subject":"Update master for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/1ae346824c3583f5ab701e3d0bc073542c3ed91a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-07 16:22:20.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-08 11:18:46.000000000","tz":60},"subject":"bindep: Use new mysql-* package names","message":"bindep: Use new mysql-* package names\n\nDebian Bookworm switched from MySQL to MariaDB.\n\nChange-Id: Id5f64a5cedd45769188be12fb5df8e43b2abcb1d\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/5854e6b0bb3ab47ace8467c2c6da659c5f31af07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/5854e6b0bb3ab47ace8467c2c6da659c5f31af07"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
