)]}'
{"id":"openstack%2Foslo.db~895625","triplet_id":"openstack%2Foslo.db~stable%2F2023.2~Ie3f213f30897cf6596f0118642d5c3cc638734dd","project":"openstack/oslo.db","branch":"stable/2023.2","topic":"revert-oslo.db-sqla2-bobcat","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-09-18 10:04:08.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"Ie3f213f30897cf6596f0118642d5c3cc638734dd","subject":"Revert \"Get test suite to full pass with SQLAlchemy 2.0\"","status":"ABANDONED","created":"2023-09-18 09:20:39.000000000","updated":"2023-09-19 05:26:41.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c2525cff68860a0606ba869dcfdc5453ebe3d873","_number":895625,"virtual_id_number":895625,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-09-18 10:25:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2023-09-18 10:04:08.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-18 09:42:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b04c003ee924ac8bd78779d59d7cce469cd6c6b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-18 09:20:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c6e49d5d6dc763888121f323f602c639b2891079","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-18 09:21:55.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"1fd978f4fdd983f4c89582023258a720cd340f93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 09:42:40.000000000","message":"Patch Set 1: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cca9a515affb4fc09c2ded7004636d23\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed06d3523c964ace9fc812df9e264cb4 : FAILURE in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b0e6425ac9c47978bd8d43cc6ae20aa : FAILURE in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d7cba4cb2a14018a2f6dc5fca4af091 : FAILURE in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/484ff2454ec442e7a011a8c1af571095 : FAILURE in 3m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a79af124038b4c818708dee67f43b44d : RETRY_LIMIT in 2m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be89fe3fd7754c5981d1826b89cfb9c6 : FAILURE in 3m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ec7f5c4a0b144f5ae618576b183e53c : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":1},{"id":"e690d051118184c04e037f80bb14b626165167a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-18 10:02:36.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d9290e96a2bf6b661d047a9e2a4e02b43c7e4a89","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-18 10:04:08.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"6099a851c70f920b36b2f25fa19ab1b49d195f77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 10:25:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00251be98906457897728916ff668005\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/803801226b0b46639ace9085c3beaf05 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/349d85b7df7c462ba61c5529fedfca2d : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee0a25fbd71346eabafaaa267df7a18a : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6815d653a93a4c65bc2676776c96ea1a : SUCCESS in 4m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d1595542b6fc4d75938ab560e576eced : RETRY_LIMIT in 2m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34ab49ebc4464f96acf1ea52c878e8d5 : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0186426acb8e418ca686eaa1a7cb71c4 : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":2},{"id":"c2525cff68860a0606ba869dcfdc5453ebe3d873","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-09-19 05:26:41.000000000","message":"Abandoned\n\nthe stack at https://review.opendev.org/c/openstack/oslo.db/+/895632 is a more subtle solution","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3c8f513762584361badff5ad27a2a96e75c0bef2","revisions":{"94d1e4ad9071be67999dbdb416d7071bdbfab91b":{"kind":"REWORK","_number":1,"created":"2023-09-18 09:20:39.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/25/895625/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/25/895625/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/25/895625/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/25/895625/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/25/895625/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/25/895625/1"}}},"commit":{"parents":[{"commit":"3a3b700d6fa267944cf3c4d8a74134995dc69977","subject":"Add back imports in sqlalchemy/utils.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/3a3b700d6fa267944cf3c4d8a74134995dc69977"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-09-18 03:07:25.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-09-18 09:18:44.000000000","tz":120},"subject":"Revert \"Get test suite to full pass with SQLAlchemy 2.0\"","message":"Revert \"Get test suite to full pass with SQLAlchemy 2.0\"\n\nThis reverts commit 1f003bcb0b74cecbfe24f52402328b76a07c9983.\n\nChange-Id: Ie3f213f30897cf6596f0118642d5c3cc638734dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/94d1e4ad9071be67999dbdb416d7071bdbfab91b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/94d1e4ad9071be67999dbdb416d7071bdbfab91b"}]},"branch":"refs/heads/stable/2023.2"},"3c8f513762584361badff5ad27a2a96e75c0bef2":{"kind":"REWORK","_number":2,"created":"2023-09-18 10:02:36.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/25/895625/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/25/895625/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/25/895625/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/25/895625/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/25/895625/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/25/895625/2"}}},"commit":{"parents":[{"commit":"c61df5e885919e95d23e43e6fa595281e395aeea","subject":"Revert \"Remove sqlalchemy-migrate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/c61df5e885919e95d23e43e6fa595281e395aeea"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-09-18 03:07:25.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2023-09-18 10:02:14.000000000","tz":120},"subject":"Revert \"Get test suite to full pass with SQLAlchemy 2.0\"","message":"Revert \"Get test suite to full pass with SQLAlchemy 2.0\"\n\nThis reverts commit 1f003bcb0b74cecbfe24f52402328b76a07c9983.\n\nAmended to drop code in oslo_db/sqlalchemy/utils.py related to NBD.\n\nChange-Id: Ie3f213f30897cf6596f0118642d5c3cc638734dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/3c8f513762584361badff5ad27a2a96e75c0bef2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/3c8f513762584361badff5ad27a2a96e75c0bef2"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
