)]}'
{"id":"openstack%2Foslo.db~895626","triplet_id":"openstack%2Foslo.db~master~Ib49ed6886470284e6486581290af4e0f2482e5c5","project":"openstack/oslo.db","branch":"master","topic":"readd-autocommit","hashtags":[],"change_id":"Ib49ed6886470284e6486581290af4e0f2482e5c5","subject":"Temporarily re-add autocommit support","status":"ABANDONED","created":"2023-09-18 09:22:46.000000000","updated":"2023-09-18 11:55:56.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"defc0a6c3411542a8f90793a5ac27cf664fa7049","_number":895626,"virtual_id_number":895626,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-09-18 11:03:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":-2,"date":"2023-09-18 10:00:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-18 11:03:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"044983412f45dde90866e9163de22ae7e64f65e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-18 09:22:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1fe514ca69332325353d324b086e45b92fe60236","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-18 09:31:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d692e6b29677006fe8173c21c5b4a9397289a2f7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-18 10:00:23.000000000","message":"Patch Set 2: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d20527d536ba12a644c9db82eeb2db99a048798e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 11:03:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec08c1e522c04db69400374447dd8980\n\n- oslodb-tox-py38-tips https://zuul.opendev.org/t/openstack/build/508f232470884745be5256eb3d1e87d2 : SUCCESS in 6m 47s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/684ecdb28de34ce1877c1eb4270e4f0b : SUCCESS in 9m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/028e74aa175e403381c79fec143f029d : SUCCESS in 1h 27m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/029074f4dc9246988e90cc5ca78f4593 : SUCCESS in 7m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3a3e3ee9b1e402f80658c6253e243c7 : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a2a33837ce043359fd6962400d379ef : SUCCESS in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a004b91b1ad4443a3ce1cc5dd77614e : SUCCESS in 4m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a8cf8f5856bb4a89b3ade870bfe4f877 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/299183118f344ec986343234929e415e : SUCCESS in 4m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aecdc7e0c7e043feb6dafe08e9d3c6a6 : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":2},{"id":"defc0a6c3411542a8f90793a5ac27cf664fa7049","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-18 11:55:56.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f9d5c047bf1be810d592902ad582ccdba9d86e71","revisions":{"6d8e7f8f1cec58965b608cf73754a0a304b576ce":{"kind":"REWORK","_number":1,"created":"2023-09-18 09:22:46.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/26/895626/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/26/895626/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/26/895626/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/26/895626/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/26/895626/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/26/895626/1"}}},"commit":{"parents":[{"commit":"5854e6b0bb3ab47ace8467c2c6da659c5f31af07","subject":"bindep: Use new mysql-* package names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/5854e6b0bb3ab47ace8467c2c6da659c5f31af07"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-18 09:05:25.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-18 09:22:37.000000000","tz":60},"subject":"Temporarily re-add autocommit support","message":"Temporarily re-add autocommit support\n\nThere are a number of projects that have yet to migrate away from\nautocommit. To prevent them blocking the release, temporarily re-add\nautocommit support. We can revert this patch early in C.\n\nThis is effectively a partial revert of change\nIfaca67c07f008d8bc0febeecd3e200cc7ee7a4b0 but focused solely on\nautocommit and with additional logic in tests to prevent us trying to\nuse autocommit with SQLAlchemy 2.x (where it\u0027s not supported).\n\nChange-Id: Ib49ed6886470284e6486581290af4e0f2482e5c5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/6d8e7f8f1cec58965b608cf73754a0a304b576ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/6d8e7f8f1cec58965b608cf73754a0a304b576ce"}]},"branch":"refs/heads/master"},"f9d5c047bf1be810d592902ad582ccdba9d86e71":{"kind":"REWORK","_number":2,"created":"2023-09-18 09:31:07.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/26/895626/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/26/895626/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/26/895626/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/26/895626/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/26/895626/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/26/895626/2"}}},"commit":{"parents":[{"commit":"5854e6b0bb3ab47ace8467c2c6da659c5f31af07","subject":"bindep: Use new mysql-* package names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/5854e6b0bb3ab47ace8467c2c6da659c5f31af07"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-18 09:05:25.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-18 09:28:11.000000000","tz":60},"subject":"Temporarily re-add autocommit support","message":"Temporarily re-add autocommit support\n\nThere are a number of projects that have yet to migrate away from\nautocommit. To prevent them blocking the release, temporarily re-add\nautocommit support. We can revert this patch early in C.\n\nThis is effectively a partial revert of change\nIfaca67c07f008d8bc0febeecd3e200cc7ee7a4b0 but focused solely on\nautocommit and with additional logic in tests to prevent us trying to\nuse autocommit with SQLAlchemy 2.x (where it\u0027s not supported).\n\nChange-Id: Ib49ed6886470284e6486581290af4e0f2482e5c5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f9d5c047bf1be810d592902ad582ccdba9d86e71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f9d5c047bf1be810d592902ad582ccdba9d86e71"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
