)]}'
{"id":"openstack%2Foslo.db~895630","triplet_id":"openstack%2Foslo.db~master~Icdb6a83be28e731b3a99df88b20d27829bcf50be","project":"openstack/oslo.db","branch":"master","topic":"readd-autocommit","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-09-18 11:56:01.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Icdb6a83be28e731b3a99df88b20d27829bcf50be","subject":"Temporarily re-add migration helpers, tests","status":"ABANDONED","created":"2023-09-18 09:58:25.000000000","updated":"2023-09-18 11:56:01.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"55992626b1e1049f54d733eb04239c7197ef41cc","_number":895630,"virtual_id_number":895630,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-09-18 11:07:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":-2,"date":"2023-09-18 10:00:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"date":"2023-09-18 10:00:29.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-18 11:07:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1bedaa2b854213569f38467ca2c4e74178076dbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-18 09:58:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55a6aceaf41c807deb3be07f969ef02010af727b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-18 10:00:16.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c9c6185111f682f81c3cbf56d191d6a370d4f65f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-18 10:00:29.000000000","message":"Patch Set 1: Code-Review-2 -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"c40ea21cc5f4b81e8e73199d08daf7e9b213601a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 11:07:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8441873ad36643d5aa971f75f91686c8\n\n- oslodb-tox-py38-tips https://zuul.opendev.org/t/openstack/build/ef833caf5c2b491381f94608fe8f7f92 : SUCCESS in 5m 47s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/9deeabde5a2146cbafcc8a51e5cc8fe8 : SUCCESS in 6m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d4a3671ea9e34ab3bb5aeacb9b7bc507 : SUCCESS in 1h 07m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/786bc5e0f6a2454dac0e8bf89553ed7c : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68fcf8e2211443dc8eb6174c6f358ca0 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0f1cf9c18144fe9a800bc86355bc400 : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/54b9ddc68ac44d73926a313c67223df3 : SUCCESS in 5m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e668997695c64d88bed0d822037f8ae6 : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cea410930a0410e853eff64ab8dbeb5 : FAILURE in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/384fc24a7e5140efaacdcdab611c13b0 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":1},{"id":"55992626b1e1049f54d733eb04239c7197ef41cc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-18 11:56:01.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a5cae82e964ce4e6d0b4996de80a7c00e39802c4","revisions":{"a5cae82e964ce4e6d0b4996de80a7c00e39802c4":{"kind":"REWORK","_number":1,"created":"2023-09-18 09:58:25.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/30/895630/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/30/895630/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/30/895630/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/30/895630/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/30/895630/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/30/895630/1"}}},"commit":{"parents":[{"commit":"f9d5c047bf1be810d592902ad582ccdba9d86e71","subject":"Temporarily re-add autocommit support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f9d5c047bf1be810d592902ad582ccdba9d86e71"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-18 09:32:25.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-18 09:56:56.000000000","tz":60},"subject":"Temporarily re-add migration helpers, tests","message":"Temporarily re-add migration helpers, tests\n\nThere are a number of projects that have yet to migrate from\nSQLAlchemy-Migrate to Alembic and are relying on some old\nSQLAlchemy-Migrate helpers. To prevent them blocking the release,\ntemporarily re-add the helpers (though not the tests for same, since\nthose won\u0027t pass with SQLAlchemy 2.x and would necessitate re-adding\nSQLAlchemy-Migrate as a dependency). We can revert this patch early in\nC.\n\nThis is effectively a partial revert of change\nIc3d6bd318038d723b0d50d39e45f8e26289e9a57 but it only re-adds the test\nhelper rather than the entirety of oslo.db\u0027s former SQLAlchemy-Migrate\nsupport.\n\nChange-Id: Icdb6a83be28e731b3a99df88b20d27829bcf50be\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a5cae82e964ce4e6d0b4996de80a7c00e39802c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a5cae82e964ce4e6d0b4996de80a7c00e39802c4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
