)]}'
{"id":"openstack%2Foslo.db~917804","triplet_id":"openstack%2Foslo.db~master~I5a068436ec1321d7a1aa39cfdd69743ea6ef4064","project":"openstack/oslo.db","branch":"master","topic":"excludes-cleanup","hashtags":[],"change_id":"I5a068436ec1321d7a1aa39cfdd69743ea6ef4064","subject":"Remove old excludes","status":"MERGED","created":"2024-05-01 02:30:58.000000000","updated":"2024-05-01 07:49:40.000000000","submitted":"2024-05-01 07:48:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"917804-excludes-cleanup","meta_rev_id":"b1a8c245cd6fe94a58a6f06f2dd48de7dbfdf986","_number":917804,"virtual_id_number":917804,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-01 07:48:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-05-01 06:40:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-05-01 06:40:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-01 04:00:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"481df5c2288dd130a9ab27811ed0fe4e5da306d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-01 02:30:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b24a0d552cec3fe261ddc51a070c41208e66aae9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 04:00:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0ed31a3bd0e4610a7ca0a5e6125bd39\n\n- oslodb-tox-py38-tips https://zuul.opendev.org/t/openstack/build/c5b54723c2514488a8fd94b380aa0f2e : SUCCESS in 6m 36s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/f240a71ffb5c4ccc95e5c4d618631d09 : SUCCESS in 8m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/608dd7ebfbb34548b0c53c710f0870ac : SUCCESS in 2m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/703b6440a96f413a8c8b04fb5217e6f5 : SUCCESS in 1h 27m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea6bfb36bdc94341a9200e75e58b5977 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/591801c5c34b48bf8163ce148b7a95dc : SUCCESS in 4m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/880f41e8d06344d3acbfc605f9705c5a : SUCCESS in 4m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/476a43167f854c3ea18e2bb37a4772a6 : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e561477ac8044b0900f955c9d36447b : SUCCESS in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a4db987416d4b5fb160806d08344102 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":1},{"id":"e684f9542f2ade1f62554fe67d99e2f72a139079","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-01 06:40:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"30a7127458d860157ae3547c4e78040dee887398","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 06:41:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ecd4ad817c46301ecae3e62a8c24debc37887da7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 07:48:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/052dccc084bb4fceb2c00288c5531a05\n\n- oslodb-tox-py38-tips https://zuul.opendev.org/t/openstack/build/b05c36604e964f8889afb7d30bb6b9eb : SUCCESS in 5m 27s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/0c2dc4740a624b3dadf45f0bbfd12227 : SUCCESS in 6m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3fa7677b9749431283ce86c623768935 : SUCCESS in 2m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4ab9bf9263ed4b86996d4d3bea00af23 : SUCCESS in 1h 07m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe8096f0d4e844b5b73f5e9d3a4c9bec : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d461306da534465b45156a5865eea17 : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ae745ae865fa44b692c7e2b8923ab199 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86c834346f5346028bb67f602324565e : SUCCESS in 3m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c29d6b16fa541c79226f1d9d367e20d : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":1},{"id":"2763a9a9a1880a4e3940efa8fe13d59b9b583942","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 07:48:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b1a8c245cd6fe94a58a6f06f2dd48de7dbfdf986","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 07:49:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c15c93b48b3240ac9c8fea8e49b51c98\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0c6d7d021a34665b26a0281fa7a7eae : SUCCESS in 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f91060aed5394b67bcdc47694ea51c03 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"14ab6451d0d3b4578c82af98dff5016424bc22cf","revisions":{"14ab6451d0d3b4578c82af98dff5016424bc22cf":{"kind":"REWORK","_number":1,"created":"2024-05-01 02:30:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/04/917804/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/04/917804/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/04/917804/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/04/917804/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/04/917804/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/04/917804/1"}}},"commit":{"parents":[{"commit":"8e1f869910e2fee3d567565b28b9cd520b07bacf","subject":"exc_filters: Handle OperationalError for MariaDB/Galera","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/8e1f869910e2fee3d567565b28b9cd520b07bacf"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-01 02:30:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-01 02:30:49.000000000","tz":540},"subject":"Remove old excludes","message":"Remove old excludes\n\nThese are detected as errors since the clean up was done[1] in\nthe requirements repository.\n\n[1] 314734e938f107cbd5ebcc7af4d9167c11347406\n\nChange-Id: I5a068436ec1321d7a1aa39cfdd69743ea6ef4064\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/14ab6451d0d3b4578c82af98dff5016424bc22cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/14ab6451d0d3b4578c82af98dff5016424bc22cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
