)]}'
{"id":"openstack%2Foslo.db~922976","triplet_id":"openstack%2Foslo.db~master~I19436e57b2e9954e25bdbece48b563c4e708a8a6","project":"openstack/oslo.db","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-12-10 10:59:20.000000000","reason":"Change was submitted"},"11816":{"account":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"last_update":"2024-12-10 10:59:20.000000000","reason":"Change was submitted"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2024-12-10 10:59:20.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2024-12-04 09:04:44.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2024-12-10 09:18:42.000000000","reason":"\u003cGERRIT_ACCOUNT_31245\u003e replied on the change","reason_account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}}},"hashtags":[],"change_id":"I19436e57b2e9954e25bdbece48b563c4e708a8a6","subject":"Create new asyncio version of enginefacade","status":"MERGED","created":"2024-06-27 19:30:25.000000000","updated":"2024-12-10 11:00:37.000000000","submitted":"2024-12-10 10:59:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":42,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"922976-eventlet-removal","meta_rev_id":"a15ddc9114e9a833036374a99c0586a820cf455d","_number":922976,"virtual_id_number":922976,"owner":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-10 10:59:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":2,"date":"2024-12-04 09:04:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-12-10 09:18:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-12-10 09:18:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-27 20:56:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-28 12:21:54.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2024-06-28 17:09:35.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2024-08-20 14:18:17.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2024-08-20 14:32:03.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-09-27 08:28:31.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2024-10-16 13:14:07.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"CC"},{"updated":"2024-10-22 14:03:20.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-12-10 09:18:42.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"}],"messages":[{"id":"2e88ceffad92baeedc2befb46b5bee40e619f504","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-06-27 19:30:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9cc377d54498053edfb90ec013135efaaf5fbe7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 20:56:49.000000000","message":"Patch Set 1: Verified-1\n\n(48 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8d897ee8a784f1b9c4892d22cfd5bcc\n\n- oslodb-tox-py38-tips https://zuul.opendev.org/t/openstack/build/be9b63bd298e42cdbcd8f727161107d8 : TIMED_OUT in 41m 02s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/f1164efbfc814064b73bd69fc10534ce : TIMED_OUT in 40m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7c14b4ea442c45879a2372e9603347b7 : SUCCESS in 1h 24m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5b23096bd584740ad94092c41710d8e : FAILURE in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/667dc412d4c3471c825c2c72c2c512c1 : TIMED_OUT in 40m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5812caa8483d40a9b4ec02449286394b : TIMED_OUT in 41m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5b940983aa5348ec8670447a9a11b6e3 : TIMED_OUT in 41m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cf5b5bb85644778ad31241484935bcd : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":1},{"id":"831d1e7fab7af061a0064fafdb4c989d01fa7996","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-06-28 16:16:55.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"da747aff8adfdaedf22688aacaedb67943ade1b8","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-06-28 17:09:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3ef941b270c684afbf2e4ecef29a316efc97d9a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-28 17:38:35.000000000","message":"Patch Set 2: Verified-1\n\n(44 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/79822a6d251c4859bd947075facb692a\n\n- oslodb-tox-py38-tips https://zuul.opendev.org/t/openstack/build/c4207169676145a6b70b37bd52bd3bb4 : TIMED_OUT in 41m 12s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/5bc85e93a0564edab15042d2422b9fbb : TIMED_OUT in 40m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fe7581bdc02b49c492b916fcbacc4016 : SUCCESS in 1h 21m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98c9a9e8ad584d9e97136361463fbd66 : FAILURE in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/20626cb0dbec4f66bc42d59b4fb516b4 : TIMED_OUT in 40m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0632278b5a5f49b1bc0a517077834df2 : TIMED_OUT in 40m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2f2eb17e25ed4d0cbf332800834df2a8 : TIMED_OUT in 40m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1dde8fc79f340b29c06892da84ff05c : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":2},{"id":"28ed3aa4081db60f6d999e65b253a2b2e4efa05d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-07-01 17:28:11.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"200799e9c00b6049d74e146b6ae737b7776da3b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-07-01 18:23:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a31dc845ceb1339b18170ab78ed3a5fbc382729c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-01 20:15:10.000000000","message":"Patch Set 4: Verified-1\n\n(40 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/333cd93d109f4a5b84b00ddd707e6338\n\n- oslodb-tox-py38-tips https://zuul.opendev.org/t/openstack/build/cc0f7626b4e4434790cd051abb8b6660 : TIMED_OUT in 40m 48s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/fc26bed00cb44836b28090f480f1399d : TIMED_OUT in 40m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6ad309f662504aa098a0a3273cf62698 : FAILURE in 2m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b1b0594753c048589f2059012e64b68d : SUCCESS in 1h 41m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88aec0609c5c4654aceb3570b109533b : FAILURE in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e9b78a6e73448ed9d14009e8856b586 : TIMED_OUT in 40m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c1bc848892814dceb638c136e4757325 : TIMED_OUT in 40m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/adb287f247d546b9afac3e02eae17be7 : TIMED_OUT in 40m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d34172a672f34dd9b1dc542a0c24e5e7 : SUCCESS in 4m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a3a4bbf8597499b9acb00d2ce98924c : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":4},{"id":"9e735b6e2dc2532fb7066c314f32fce56a613d05","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-08-29 14:06:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ba58e6ddd5b37a3f7c8f751491a4f1018bb27a7e","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-08-29 14:12:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c8148dc89c06e81aeff14bed3682c8c9f28ce4c1","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-08-29 14:13:04.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"eaa8872ce1e0d6b4310fe03694f4eae6cd5afe1b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-09-18 07:26:18.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fce505a6c95496d7c65ad4b30eb2f8697656ae42","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-09-23 14:56:30.000000000","message":"Patch Set 4:\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"c30e1132c43f678104876167b7580ef01d9f8382","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-09-23 15:07:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cfa0927623a6ca0ba131b571cd7ef3d3099d475e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-09-25 13:15:30.000000000","message":"Patch Set 4: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"66421a5f5d86534658f579471a952fe7c9bca7b6","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-09-25 13:17:48.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d9110e8129bc1561b1117141d0021bc13a4ccee9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-09-26 15:12:04.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"312d37502a10b824e854d7749559fb7e23d64d65","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-09-26 15:12:10.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"a91052f8c412863ebf908aef0b114134d68af224","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 16:59:09.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/15a7c4360d0545a0875f067dfad3f850\n\n- oslodb-tox-py38-tips https://zuul.opendev.org/t/openstack/build/b407b35396b84b6aab21184394b70952 : TIMED_OUT in 40m 49s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/82cc07d9dc134d7190adc3f07b39c12b : TIMED_OUT in 40m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0373fc6497bb4ceca0ee088b0f80f8f2 : FAILURE in 2m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9d4efaa670f14d149e7e42a6ac75e290 : SUCCESS in 1h 25m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87433045f4644efd9b149c3c29ff24ca : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6a4c11ababe4551aebf8f25d5543708 : TIMED_OUT in 40m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f90f14dac3c247688e41e7c86c39842a : TIMED_OUT in 41m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b43d4fe7f79c4838a3e47d5072e508f1 : TIMED_OUT in 40m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bf6a3cb4ff74573a54593b9934c89f9 : SUCCESS in 3m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2c162f255904c138fb273751221d892 : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":5},{"id":"1eaaf1fafd748e531d527231a3446b15c2d999a8","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-09-30 20:27:46.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8a142684951e0816d47ff3f7fcd022b8df8e2bb8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-10-01 08:54:59.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"e780165de57edad763ab155d3e74dc763a956a6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-10-01 17:37:44.000000000","message":"Uploaded patch set 6: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"c9d7c77730218391fd366715b53237c54dfcdf7d","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-10-01 17:38:01.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"ced873268fbf99abe6070c221a69ad08ffe77a63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 19:20:32.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6bc07ea10bfe46d5b3bf03581e5ad80b\n\n- oslodb-tox-py38-tips https://zuul.opendev.org/t/openstack/build/40fe26b51dde4c65aae52fdd938c9cf8 : TIMED_OUT in 41m 14s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/c35abab88eac4595a752c40cdf20b5f2 : TIMED_OUT in 41m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8762f0b70913476ab76ab970e940fc91 : SUCCESS in 3m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/36c0096ef96743b4a24501e58c60f295 : SUCCESS in 1h 37m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d487f3cd3a84053a9927d249626f748 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84e7b6a6ca2648519e505fff3c997016 : TIMED_OUT in 40m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/56a52471e24f49c1b8024e27d74b4fb9 : TIMED_OUT in 40m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5ea1701f1fb4b8f981601f554fefd9f : SUCCESS in 4m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d5e12d00a4e4fe6a297c91b5b4d67a3 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":6},{"id":"d944c4177d5690cc659e796b8d6e6f9c787e04ee","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-10-14 09:07:44.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"01edd3d8db7f67a3cce4a5bf98fbfa1c79750974","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-14 10:54:59.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b4e7b92cadf4273adec6833db55ee45\n\n- oslodb-tox-py38-tips https://zuul.opendev.org/t/openstack/build/c25b537f5e9f4563896e09897d669828 : TIMED_OUT in 41m 13s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/f06a31b84b4c4d13aeb43848689c214d : TIMED_OUT in 40m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b359220e4456483a9b9c2ddf308205b3 : SUCCESS in 2m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ac91c86f69c64d17bc3da06f9d73095d : SUCCESS in 1h 42m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c55cb8e39e6945969fd3a49ab41780f6 : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6010d58cc35f4d5a859da7e88c37cc27 : TIMED_OUT in 40m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e2e90c24844a4ebfa3fc4ccb49d9ca47 : TIMED_OUT in 40m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5de33f16aecf4c50829c70f26c58eaaa : SUCCESS in 4m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba21501043c84ea7bc20fc679eb10dfa : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":6},{"id":"74ce85fb5dd72b4f391234884f30446d561208e4","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-10-16 13:14:07.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0a9be39304dec5072c51cc162dcc9134cf8d81f4","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-10-16 13:17:13.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"29860c52378c6daa2e8a379aabed05d0d41a7f64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 14:48:44.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5ce1513ff48849a3990656d32805e36d\n\n- oslodb-tox-py38-tips https://zuul.opendev.org/t/openstack/build/7998a8f7b0b046cbac612d7d4f5b3fc7 : TIMED_OUT in 40m 51s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/2e77df76315c4473aa2e823dc6b630cd : TIMED_OUT in 40m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/97a023aa2d324d68a788561004668e3b : SUCCESS in 3m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ae796999ae1f4d2f91b4f6882928d0f6 : SUCCESS in 1h 32m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c1ade1abb5f46d698e79a0bd3bab46a : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc2e6d7932bd4ef6b44edb1f07d67ddf : TIMED_OUT in 40m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e77fa6e6cbc34e559fce760669a1b41f : TIMED_OUT in 40m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1af2aeaf76674c63b7bf9fd707b8a721 : SUCCESS in 4m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a85ab277357454ba272f6908ca612c3 : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":6},{"id":"f0166aba881ccd315b61875e9c74ba69c629ee44","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-10-24 08:17:03.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"42cffa72a573ec224b92e50b97a0b3fba68d3237","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 09:56:54.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a84e0ee6ea84434b90f6eb71dfe48f3\n\n- oslodb-tox-py39-tips https://zuul.opendev.org/t/openstack/build/fa197b50d7064f9a9a6289852395680b : TIMED_OUT in 40m 44s\n- oslodb-tox-py312-tips https://zuul.opendev.org/t/openstack/build/7b88c2b6012544bd8104a94d5626564b : TIMED_OUT in 41m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f8f54a077d9d4dbd9f91b03761724aaf : SUCCESS in 2m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/02460be2d8644e058a7c817994126573 : SUCCESS in 1h 38m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0c6fd2207f447768acce21361124be4 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/defa5ad31607492d9b3c933583d7bc54 : TIMED_OUT in 40m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cc136df5cfa44cea8e91c16d35e68c6d : TIMED_OUT in 40m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a149b9e01c4c4dd6b7b4b81f012c48be : SUCCESS in 2m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f66d2b46edbf4c829aa84cdcfa9bd6c9 : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":6},{"id":"cf0f92124033719b52c53667bbe9cc590b9fbebb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-11-05 14:09:23.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d9e41e04ec0a5eadfcae6a99e2878c8b0031a5e3","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-11-05 14:36:34.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"87f405db16d8ef5a2ec6029af26f4995887683a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-11-05 14:38:01.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"a7ea6c58c6aafcba014da5dfc698f9f806bc231e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 16:23:02.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b1684563fd7e48e09eb041b8059c53ea\n\n- oslodb-tox-py39-tips https://zuul.opendev.org/t/openstack/build/97bc3be489a744ac9d879e5f735429c9 : TIMED_OUT in 40m 53s\n- oslodb-tox-py312-tips https://zuul.opendev.org/t/openstack/build/5e1827ffdd2e4f8aa215dfc0defbdd8c : TIMED_OUT in 40m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9de7113a4ac44a82b40edc502b7f25e1 : SUCCESS in 2m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0bd8303d126a4f089323dc0c91bb0b7f : SUCCESS in 1h 36m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/651e078beb49448f8c459e578da526cc : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2b260e36e8246abb5cb2283556c0a59 : TIMED_OUT in 40m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/85a5a52aee944a369c1c5f13a0c20714 : TIMED_OUT in 40m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f560202655e4fa984d567cc6c09c434 : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8f4f1ab4532483997f4f053f6be8968 : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":7},{"id":"eb2c26e79267cd14cb3ae5ab5733a8dcc3559347","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-11-07 13:05:38.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"0ad7c33296f4441a5025926031d50f448cbed589","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 14:28:13.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a9ecee5ccaf048bd9cdc85b387351bac\n\n- oslodb-tox-py39-tips https://zuul.opendev.org/t/openstack/build/0a5faf62a74f43ecbfcb31cc3364575e : TIMED_OUT in 40m 46s\n- oslodb-tox-py312-tips https://zuul.opendev.org/t/openstack/build/0196b783f142463f8380bd89c59f3e72 : TIMED_OUT in 40m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1567e9f080304bd1a433abfeba1e3572 : SUCCESS in 2m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/168719bd691247d2a824adbe122264c5 : SUCCESS in 1h 21m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b57cd839f114675ba8211fd4cedbc14 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34adf7667a70452ca9cd71e429dca429 : TIMED_OUT in 40m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e735881ab0e74016bec8f287e9cf182a : TIMED_OUT in 40m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16286ca6cd20463392733060a5296e8f : SUCCESS in 3m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fbc7b4bf1ebc4371a54c8e071dfaaca8 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":7},{"id":"54e2cf54945210deafc5d7d397f8ff1c07317c21","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-11-12 10:22:35.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"19412f3a9a005517c603be41de832e2f6689e9be","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-11-12 10:24:42.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ad3d05bfb0eae34441253793efc74e5b43bb2028","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-11-28 08:00:25.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"9bc452743033bdb9553ab5a231b93bc18c42e746","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 09:04:07.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b04ea4f2f2c4614979648bbeb92db08\n\n- oslodb-tox-py39-tips https://zuul.opendev.org/t/openstack/build/5b8919d2325a43d6ab32406279cdecfe : TIMED_OUT in 40m 49s\n- oslodb-tox-py312-tips https://zuul.opendev.org/t/openstack/build/7629eb16569b4dfa833db2a87485d2b6 : TIMED_OUT in 40m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9e8e3772aaad4536939d6a100dcf812d : SUCCESS in 2m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/84c5c064857f40118e172bcfb2eaecd7 : SUCCESS in 1h 02m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a64da74e81b1473a87a586c09398bdad : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3db40e8470f4062b964e052b036cdaf : TIMED_OUT in 40m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/22a00fe202034e45bb0b65f4bf6056fc : TIMED_OUT in 40m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0de13ed17dae409a8b08b84808ed2fd3 : SUCCESS in 2m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ce10f5f82aa4746b75aff2151cc26b7 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":7},{"id":"4e2257e05dc5422f6a434db74d47643fa87c67ae","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-12-03 14:36:53.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4e7994e6c93b409e05a3494475329ae536de2144","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-12-03 15:11:12.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"2f76e01f52017db8b8206a3eb1cc7b2eaf0da3b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-12-03 15:26:28.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"d76df57dd99035454ae98b6afa673947fee5ae02","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-12-03 15:26:37.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b0dd4af25b9c9105e7ea3655dd20f08d1cbea4e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 16:49:57.000000000","message":"Patch Set 8: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0a6b36dc341547e8a7ced7ffa9affcd0\n\n- oslodb-tox-py39-tips https://zuul.opendev.org/t/openstack/build/64f80374e3d64cc2b94797a4d939212e : SUCCESS in 3m 26s\n- oslodb-tox-py312-tips https://zuul.opendev.org/t/openstack/build/3c0d7407025444088fad95ab36359445 : SUCCESS in 6m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e8f8b80f03e2404b8a9445525565571a : SUCCESS in 2m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/89d111c26939433989863535aae1d6be : SUCCESS in 1h 21m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58a18321a6f44dd5af2aaa78410d3f4a : FAILURE in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/953a637a29b643de81f083cac584797f : SUCCESS in 5m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c09c6293a8b748b79d00f0a9fe58c4f9 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0062009afd3a49b389a1f2fee4b45c13 : SUCCESS in 2m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b287357fcb07404d8bb39dcd1d7fca4c : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":8},{"id":"e7b6c06fc45382128310e140befbd89775dd1c90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-12-03 18:20:39.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"3ebbff711203d055159e720a7cf2f8e3a8abd05b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 20:25:12.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f82c475d95104c169700b9957cc76600\n\n- oslodb-tox-py39-tips https://zuul.opendev.org/t/openstack/build/fc3e449ccd7645caae930caeb9b9aa77 : SUCCESS in 4m 38s\n- oslodb-tox-py312-tips https://zuul.opendev.org/t/openstack/build/163cd96858b34f17a97b23e9f91374d5 : SUCCESS in 5m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/435f486ab4f8456a9ecc73cdef6ea309 : SUCCESS in 2m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/edafddf00ccd4ab889932af39646a368 : SUCCESS in 1h 52m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1243c8b61b0549e28d71fd445167b076 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/24067f16c5454ec3a1ead03e9e9bf53e : SUCCESS in 4m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/644d11d56b8a45e79e2d39998cb8463a : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ed1a121fe334b3fba843a36342cd074 : SUCCESS in 3m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/201069c49d404ab392d119cfa5971ac1 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":9},{"id":"ebae0cde8db0899551220d8e81f59b2b32a8ba30","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-12-04 09:04:44.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"dc7f19d111a871447803df0ba926e9b25f0e5729","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-12-10 09:18:42.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"442eba7934a38f1c9b1afbaf84aaad27728c604d","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-12-10 09:18:47.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"ade194316b0357272d078fbd03c5443bec81e6fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 09:19:22.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"085388781a48d649b283a1e8596beb1c22b2e07e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 10:59:20.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/888ed96cdea941a0bbaa2b9d6571c6da\n\n- oslodb-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d90919cf2eec4e0581b6f8929a2765f0 : SUCCESS in 7m 14s\n- oslodb-tox-py312-tips https://zuul.opendev.org/t/openstack/build/11f08f87637c4388a0efa3f402b8d9bd : SUCCESS in 6m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e2dc1e634eab4126bd6c7dde65e5840d : SUCCESS in 5m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b8f67a4814df4be4b597785d34cfa104 : SUCCESS in 1h 37m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/571720cc1d1a4d6d8b2d1f838047edc2 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37a350763eac44b88d75d1525d01b93b : SUCCESS in 4m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1764039c4f2b46d598304268e58a1ba0 : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72d928cc829e459198821b999eb11833 : SUCCESS in 3m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5003c2f75baa49d6b74fd22e8bf27453 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":9},{"id":"c372cff9d4d07080bb2cfb86036f0fd90e8b03a6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 10:59:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"a15ddc9114e9a833036374a99c0586a820cf455d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 11:00:37.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bc1554bfd6c4f7d826500c985f981a6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/681057d2b7024324a66d362d82f7ea36 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/142b0faae1614758a6c362f728466de3 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"7cfe7fc693894a3488f9c45e35250a458b93cb63","revisions":{"e481baf5c9f874f8dad111a12db7143f7ef47409":{"kind":"REWORK","_number":1,"created":"2024-06-27 19:30:25.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/76/922976/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/76/922976/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/1"}}},"commit":{"parents":[{"commit":"a59dba4479f9f391b63901b847d25a77093a460a","subject":"Merge \"Remove [database] use_tpool and TpoolDbapiWrapper\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a59dba4479f9f391b63901b847d25a77093a460a"}]}],"author":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2024-06-27 19:26:22.000000000","tz":-240},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2024-06-27 19:27:47.000000000","tz":-240},"subject":"WIP: Create new asyncio version of enginefacade","message":"WIP: Create new asyncio version of enginefacade\n\nWork on this is currently going on in a branch in my\nown repo at\nhttps://github.com/zzzeek/oslo.db/tree/eventlet-removal-work .\n\nThis change adds a new oslo_db.sqlalchemy.asyncio_enginefacade\nwhich provides for the identical patterns as enginefacade but using\nawaitables as well as the AsyncConnection / AsyncSession from\nsqlalchemy.ext.asyncio.\n\nCode in this patch is incomplete, and so far only allows a basic\n\"hello world\" to proceed.  I am putting it in openstack gerrit\nearly to illustrate the approach I\u0027m taking for this.\n\nChange-Id: I19436e57b2e9954e25bdbece48b563c4e708a8a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e481baf5c9f874f8dad111a12db7143f7ef47409"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e481baf5c9f874f8dad111a12db7143f7ef47409"}]},"branch":"refs/heads/master"},"b853302eb38c0e0aa6640380e9ce02415d6ca5e3":{"kind":"REWORK","_number":2,"created":"2024-06-28 16:16:55.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/76/922976/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/76/922976/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/2"}}},"commit":{"parents":[{"commit":"a59dba4479f9f391b63901b847d25a77093a460a","subject":"Merge \"Remove [database] use_tpool and TpoolDbapiWrapper\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a59dba4479f9f391b63901b847d25a77093a460a"}]}],"author":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2024-06-28 16:14:59.000000000","tz":-240},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2024-06-28 16:14:59.000000000","tz":-240},"subject":"WIP: Create new asyncio version of enginefacade","message":"WIP: Create new asyncio version of enginefacade\n\nWork on this is currently going on in a branch in my\nown repo at\nhttps://github.com/zzzeek/oslo.db/tree/eventlet-removal-work .\n\nThis change adds a new oslo_db.sqlalchemy.asyncio_enginefacade\nwhich provides for the identical patterns as enginefacade but using\nawaitables as well as the AsyncConnection / AsyncSession from\nsqlalchemy.ext.asyncio.\n\nCode in this patch is incomplete, and so far only allows a basic\n\"hello world\" to proceed.  I am putting it in openstack gerrit\nearly to illustrate the approach I\u0027m taking for this.\n\nChange-Id: I19436e57b2e9954e25bdbece48b563c4e708a8a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/b853302eb38c0e0aa6640380e9ce02415d6ca5e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/b853302eb38c0e0aa6640380e9ce02415d6ca5e3"}]},"branch":"refs/heads/master"},"03a05d1ad329db7ec92bf63c7f77e3f45d3defb4":{"kind":"REWORK","_number":3,"created":"2024-07-01 17:28:11.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/76/922976/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/76/922976/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/3"}}},"commit":{"parents":[{"commit":"a59dba4479f9f391b63901b847d25a77093a460a","subject":"Merge \"Remove [database] use_tpool and TpoolDbapiWrapper\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a59dba4479f9f391b63901b847d25a77093a460a"}]}],"author":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2024-06-28 16:14:59.000000000","tz":-240},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2024-07-01 17:28:06.000000000","tz":-240},"subject":"WIP: Create new asyncio version of enginefacade","message":"WIP: Create new asyncio version of enginefacade\n\nWork on this is currently going on in a branch in my\nown repo at\nhttps://github.com/zzzeek/oslo.db/tree/eventlet-removal-work .\n\nThis change adds a new oslo_db.sqlalchemy.asyncio_enginefacade\nwhich provides for the identical patterns as enginefacade but using\nawaitables as well as the AsyncConnection / AsyncSession from\nsqlalchemy.ext.asyncio.\n\nCode in this patch is incomplete, and so far only allows a basic\n\"hello world\" to proceed.  I am putting it in openstack gerrit\nearly to illustrate the approach I\u0027m taking for this.\n\nChange-Id: I19436e57b2e9954e25bdbece48b563c4e708a8a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/03a05d1ad329db7ec92bf63c7f77e3f45d3defb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/03a05d1ad329db7ec92bf63c7f77e3f45d3defb4"}]},"branch":"refs/heads/master"},"8d9aea0055501e407c33276669720b8a6c461004":{"kind":"REWORK","_number":4,"created":"2024-07-01 18:23:24.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/76/922976/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/76/922976/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/4"}}},"commit":{"parents":[{"commit":"a59dba4479f9f391b63901b847d25a77093a460a","subject":"Merge \"Remove [database] use_tpool and TpoolDbapiWrapper\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a59dba4479f9f391b63901b847d25a77093a460a"}]}],"author":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2024-06-28 16:14:59.000000000","tz":-240},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2024-07-01 18:23:10.000000000","tz":-240},"subject":"WIP: Create new asyncio version of enginefacade","message":"WIP: Create new asyncio version of enginefacade\n\nThis change adds a new oslo_db.sqlalchemy.asyncio_enginefacade\nwhich provides for the identical patterns as enginefacade but using\nawaitables as well as the AsyncConnection / AsyncSession from\nsqlalchemy.ext.asyncio.\n\nChange-Id: I19436e57b2e9954e25bdbece48b563c4e708a8a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/8d9aea0055501e407c33276669720b8a6c461004"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/8d9aea0055501e407c33276669720b8a6c461004"}]},"branch":"refs/heads/master"},"5c33d6f525a4ae8af90371b625b030f69d95ccdd":{"kind":"REWORK","_number":5,"created":"2024-09-26 15:12:04.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/76/922976/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/76/922976/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/5"}}},"commit":{"parents":[{"commit":"a59dba4479f9f391b63901b847d25a77093a460a","subject":"Merge \"Remove [database] use_tpool and TpoolDbapiWrapper\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a59dba4479f9f391b63901b847d25a77093a460a"}]}],"author":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2024-06-28 16:14:59.000000000","tz":-240},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2024-09-26 15:11:33.000000000","tz":-240},"subject":"WIP: Create new asyncio version of enginefacade","message":"WIP: Create new asyncio version of enginefacade\n\nThis change adds a new oslo_db.sqlalchemy.asyncio_enginefacade\nwhich provides for the identical patterns as enginefacade but using\nawaitables as well as the AsyncConnection / AsyncSession from\nsqlalchemy.ext.asyncio.\n\nChange-Id: I19436e57b2e9954e25bdbece48b563c4e708a8a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/5c33d6f525a4ae8af90371b625b030f69d95ccdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/5c33d6f525a4ae8af90371b625b030f69d95ccdd"}]},"branch":"refs/heads/master"},"389aa64dfd23f4e9ee2668af9cb1e3d9c588fa05":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2024-10-01 17:37:44.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/76/922976/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/76/922976/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/6"}}},"commit":{"parents":[{"commit":"a59dba4479f9f391b63901b847d25a77093a460a","subject":"Merge \"Remove [database] use_tpool and TpoolDbapiWrapper\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a59dba4479f9f391b63901b847d25a77093a460a"}]}],"author":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2024-06-28 16:14:59.000000000","tz":-240},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2024-10-01 17:37:15.000000000","tz":-240},"subject":"Create new asyncio version of enginefacade","message":"Create new asyncio version of enginefacade\n\nThis change adds a new oslo_db.sqlalchemy.asyncio_enginefacade\nwhich provides for the identical patterns as enginefacade but using\nawaitables as well as the AsyncConnection / AsyncSession from\nsqlalchemy.ext.asyncio.\n\nDepends-on: https://review.opendev.org/c/openstack/requirements/+/930947\nChange-Id: I19436e57b2e9954e25bdbece48b563c4e708a8a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/389aa64dfd23f4e9ee2668af9cb1e3d9c588fa05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/389aa64dfd23f4e9ee2668af9cb1e3d9c588fa05"}]},"branch":"refs/heads/master"},"0b68eb1125c9c6cb71597c78a737ff9a433659c9":{"kind":"REWORK","_number":7,"created":"2024-11-05 14:38:01.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/76/922976/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/76/922976/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/7"}}},"commit":{"parents":[{"commit":"cd19ec22c8273d3cc8cb690150ce7b7d84634077","subject":"Merge \"Remove deprecated async property\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/cd19ec22c8273d3cc8cb690150ce7b7d84634077"}]}],"author":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2024-06-28 16:14:59.000000000","tz":-240},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2024-11-05 14:37:56.000000000","tz":-300},"subject":"Create new asyncio version of enginefacade","message":"Create new asyncio version of enginefacade\n\nThis change adds a new oslo_db.sqlalchemy.asyncio_enginefacade\nwhich provides for the identical patterns as enginefacade but using\nawaitables as well as the AsyncConnection / AsyncSession from\nsqlalchemy.ext.asyncio.\n\nDepends-on: https://review.opendev.org/c/openstack/requirements/+/930947\nChange-Id: I19436e57b2e9954e25bdbece48b563c4e708a8a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/0b68eb1125c9c6cb71597c78a737ff9a433659c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/0b68eb1125c9c6cb71597c78a737ff9a433659c9"}]},"branch":"refs/heads/master"},"160d473d8c8a5e656942194d49590d5aa6de016c":{"kind":"REWORK","_number":8,"created":"2024-12-03 15:26:28.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/76/922976/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/76/922976/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/8"}}},"commit":{"parents":[{"commit":"cd19ec22c8273d3cc8cb690150ce7b7d84634077","subject":"Merge \"Remove deprecated async property\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/cd19ec22c8273d3cc8cb690150ce7b7d84634077"}]}],"author":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2024-06-28 16:14:59.000000000","tz":-240},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2024-12-03 15:26:24.000000000","tz":-300},"subject":"Create new asyncio version of enginefacade","message":"Create new asyncio version of enginefacade\n\nThis change adds a new oslo_db.sqlalchemy.asyncio_enginefacade\nwhich provides for the identical patterns as enginefacade but using\nawaitables as well as the AsyncConnection / AsyncSession from\nsqlalchemy.ext.asyncio.\n\nDepends-on: https://review.opendev.org/c/openstack/requirements/+/930947\nChange-Id: I19436e57b2e9954e25bdbece48b563c4e708a8a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/160d473d8c8a5e656942194d49590d5aa6de016c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/160d473d8c8a5e656942194d49590d5aa6de016c"}]},"branch":"refs/heads/master"},"7cfe7fc693894a3488f9c45e35250a458b93cb63":{"kind":"REWORK","_number":9,"created":"2024-12-03 18:20:39.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/76/922976/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/76/922976/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/76/922976/9"}}},"commit":{"parents":[{"commit":"cd19ec22c8273d3cc8cb690150ce7b7d84634077","subject":"Merge \"Remove deprecated async property\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/cd19ec22c8273d3cc8cb690150ce7b7d84634077"}]}],"author":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2024-06-28 16:14:59.000000000","tz":-240},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2024-12-03 18:20:35.000000000","tz":-300},"subject":"Create new asyncio version of enginefacade","message":"Create new asyncio version of enginefacade\n\nThis change adds a new oslo_db.sqlalchemy.asyncio_enginefacade\nwhich provides for the identical patterns as enginefacade but using\nawaitables as well as the AsyncConnection / AsyncSession from\nsqlalchemy.ext.asyncio.\n\nDepends-on: https://review.opendev.org/c/openstack/requirements/+/930947\nChange-Id: I19436e57b2e9954e25bdbece48b563c4e708a8a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7cfe7fc693894a3488f9c45e35250a458b93cb63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/7cfe7fc693894a3488f9c45e35250a458b93cb63"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
