)]}'
{"id":"openstack%2Foslo.db~927785","triplet_id":"openstack%2Foslo.db~master~Ib04858dda47fcbcafe027f222030506d7477e43b","project":"openstack/oslo.db","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-31 16:22:16.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ib04858dda47fcbcafe027f222030506d7477e43b","subject":"Remove deprecated async property","status":"MERGED","created":"2024-09-03 07:30:19.000000000","updated":"2024-10-31 18:08:54.000000000","submitted":"2024-10-31 18:07:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927785","meta_rev_id":"0d0306d0d6ed48fd259950fda9a50e1f1e8e76cf","_number":927785,"virtual_id_number":927785,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-31 18:07:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":2,"date":"2024-09-03 13:21:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":2,"date":"2024-10-31 16:22:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2024-09-10 07:33:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":1,"date":"2024-10-31 16:22:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-03 08:56:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-03 13:21:49.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2024-09-10 07:33:31.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"0ee18c115276b29215b00bd1e5a723caea6050de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-03 07:30:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db777009f39becbacc3c7aaa5e6075192d92e4e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-03 07:30:57.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"bd3c34c6d01a73b2b5e2f2ea1a4a9e79feceb387","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-03 07:33:30.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"5d265a14d97dd1660817583b852237d0867cfb22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 08:56:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7bd9ddaa36c4d888b1cd7e951be177e\n\n- oslodb-tox-py38-tips https://zuul.opendev.org/t/openstack/build/26aa9872a0fc487ba7f90e2d2f932960 : SUCCESS in 5m 07s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/f5b45c4de53b46738b4aa86d1c4cff2c : SUCCESS in 5m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d7e33d3f689f4b42b1be805cb5601859 : FAILURE in 1h 21m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5bcd819dd15484283b2abc6ab02c49a : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47358155c1a243e69932d4513b594e78 : SUCCESS in 5m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fbd6aad76a6b44fe8c26b1c49ef0216f : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3eef27e677194431923eaa88ee142284 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94ad12063ae24da1887dac42876efd87 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":3},{"id":"b58713433bc78cd7ffd11dfe31452da2f2eacede","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-03 09:04:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"94ef9f2cb9394dcfeb196731e82dc13b7fe464f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 10:56:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00cf62ceb8bd45489c7f6d89707ca11d\n\n- oslodb-tox-py38-tips https://zuul.opendev.org/t/openstack/build/1341f8a9004a4e9fa357f2e00ed1a468 : SUCCESS in 5m 07s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/4b8adf72e38347cc8ba6031b92da8bee : SUCCESS in 5m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/95db0c93ef6b4577ba8a93c733f29e3a : SUCCESS in 1h 34m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/614fbcd4710b4bacabead0925fe898eb : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea027651f45246a0a3c79a376fac50b4 : SUCCESS in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0a70a2be3a924cff993ad6f08869c969 : SUCCESS in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9100a0ff4e9e4f51b440ddd5949b48fa : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe41d7fe94f445b4a5424b3be33dbc9f : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":3},{"id":"4a0426236ce843850bac19dbecf4e0757d8a31de","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-09-03 13:21:49.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d29b44e8bbbafe6aeeed8249e1633bdfc5aef15c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-09-10 07:33:31.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"b529d5ba2b6686aa0ccaf7f03f56222343fd18f6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-31 16:22:16.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"11a0ddd7c529ac57f329e9c6f1a1bc92b47c457e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 16:22:47.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d1d244e95e6e16e39920e21fd3e6e2d50833358b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 18:07:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6d80507eefe42d3817c5a8a18394c66\n\n- oslodb-tox-py39-tips https://zuul.opendev.org/t/openstack/build/46a5467bb9a341128fe4de6baef350d1 : SUCCESS in 6m 07s\n- oslodb-tox-py312-tips https://zuul.opendev.org/t/openstack/build/5a4da1097cdd443bb796f92baa342974 : SUCCESS in 5m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7a9fc606b3614e648fd92a76a331b7f9 : SUCCESS in 1h 39m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6c7835eff96473fb2fa284f7849dfc2 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a39e86d622434935ab2ace6c6e36958d : SUCCESS in 4m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3631e00818fb429e9b53e3f3c3cccc3c : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/529ca822456e421eb3d0f110802128f8 : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":3},{"id":"4f0ac279130e371dbbc1fddf6e8f9bb478c1c550","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 18:07:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"0d0306d0d6ed48fd259950fda9a50e1f1e8e76cf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 18:08:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/228a230d34244520bc0687375a5dcd6d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a9da13963de449d93312ff8faa531bc : SUCCESS in 40s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a6d89f55e7b0ae298ee15b205f73587122e17821","revisions":{"ea0b93e5016f8fbe65d48e944b8b10d93aa8ad0e":{"kind":"REWORK","_number":1,"created":"2024-09-03 07:30:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/85/927785/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/85/927785/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/85/927785/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/85/927785/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/85/927785/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/85/927785/1"}}},"commit":{"parents":[{"commit":"a59dba4479f9f391b63901b847d25a77093a460a","subject":"Merge \"Remove [database] use_tpool and TpoolDbapiWrapper\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a59dba4479f9f391b63901b847d25a77093a460a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-03 07:29:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-03 07:29:40.000000000","tz":540},"subject":"Remove deprecated async property","message":"Remove deprecated async property\n\nThis no longer works in Python 3.8+ because async was made a keyword.\nSee df6bf3401266f42271627c1e408f87c71a06cef7 for further details.\n\nChange-Id: Ib04858dda47fcbcafe027f222030506d7477e43b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/ea0b93e5016f8fbe65d48e944b8b10d93aa8ad0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/ea0b93e5016f8fbe65d48e944b8b10d93aa8ad0e"}]},"branch":"refs/heads/master"},"2c855783ad699eaa9865eff08dd96e99245e185b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-09-03 07:30:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/85/927785/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/85/927785/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/85/927785/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/85/927785/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/85/927785/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/85/927785/2"}}},"commit":{"parents":[{"commit":"a59dba4479f9f391b63901b847d25a77093a460a","subject":"Merge \"Remove [database] use_tpool and TpoolDbapiWrapper\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a59dba4479f9f391b63901b847d25a77093a460a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-03 07:29:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-03 07:30:39.000000000","tz":540},"subject":"Remove deprecated async property","message":"Remove deprecated async property\n\nThis no longer works in Python 3.7 or later because async was made\na keyword. See df6bf3401266f42271627c1e408f87c71a06cef7 for further\ndetails.\n\nChange-Id: Ib04858dda47fcbcafe027f222030506d7477e43b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/2c855783ad699eaa9865eff08dd96e99245e185b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/2c855783ad699eaa9865eff08dd96e99245e185b"}]},"branch":"refs/heads/master"},"a6d89f55e7b0ae298ee15b205f73587122e17821":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-09-03 07:33:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/85/927785/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/85/927785/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/85/927785/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/85/927785/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/85/927785/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/85/927785/3"}}},"commit":{"parents":[{"commit":"a59dba4479f9f391b63901b847d25a77093a460a","subject":"Merge \"Remove [database] use_tpool and TpoolDbapiWrapper\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a59dba4479f9f391b63901b847d25a77093a460a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-03 07:29:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-03 07:32:33.000000000","tz":540},"subject":"Remove deprecated async property","message":"Remove deprecated async property\n\nThis no longer works in Python 3.7 or later because async is now\na keyword. Because now the minimum python version supported is 3.8,\nthe compatibility code is not useful.\n\nSee df6bf3401266f42271627c1e408f87c71a06cef7 for further details.\n\nChange-Id: Ib04858dda47fcbcafe027f222030506d7477e43b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a6d89f55e7b0ae298ee15b205f73587122e17821"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a6d89f55e7b0ae298ee15b205f73587122e17821"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
