)]}'
{"id":"openstack%2Foslo.db~929398","triplet_id":"openstack%2Foslo.db~master~I48c1904ddfe27d13733741b5f1ca2661b8f1d86a","project":"openstack/oslo.db","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-09-21 06:33:58.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I48c1904ddfe27d13733741b5f1ca2661b8f1d86a","subject":"Drop SQLALCHEMY_WARN_20","status":"MERGED","created":"2024-09-16 06:24:12.000000000","updated":"2024-09-21 07:51:43.000000000","submitted":"2024-09-21 07:50:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"929398-sqlalchemy-20","meta_rev_id":"91021d468c5c99904746ac06a876467dbded5a06","_number":929398,"virtual_id_number":929398,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-21 07:50:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"all":[{"value":2,"date":"2024-09-16 14:06:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":1,"date":"2024-09-21 06:33:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-16 08:04:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-16 14:06:37.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"ceaddc8c5b2905016a9d181080e485b0ad2bef69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-16 06:24:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75864f697e79003ffacc02e123b07596f378bdbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-16 08:04:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3117d2e81db5408eb37876faecf4ee3b\n\n- oslodb-tox-py38-tips https://zuul.opendev.org/t/openstack/build/b1570253b0ba44ba927fbc605820bd70 : SUCCESS in 6m 11s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/a2f8f36aa43b4a9eb1b177bf9ef1281d : SUCCESS in 6m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4920d94af8bb4a3780341ba5131bb67b : SUCCESS in 1h 38m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a392b225c48f45c6912ddaeb387c059e : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4be74b7563e6437f96d6880708234578 : SUCCESS in 4m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/139afc0f8fed4224af8f5ab86d3b2a73 : SUCCESS in 3m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2e4dcd8c98ce4569b2a10c032ca2c42e : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/309edeff70a3474d897f9ea3f033af66 : SUCCESS in 3m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a91108096c945f7802fcc4860a93053 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":1},{"id":"c13565e1be888724869454210a46516aeeae0c20","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-09-16 14:06:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9fda3e1c1ca322684092b6d396d116fddbc8767d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-21 06:33:58.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"08e0be554e5d74b34dc41c40a312197f0c248cdc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-21 06:34:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5921200398cae06926ccd04bc501861bae0f6f09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-21 07:50:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/852bbd1a093140dc9dbb1343a711181a\n\n- oslodb-tox-py38-tips https://zuul.opendev.org/t/openstack/build/bcae497f9345428f9c583f8253c96f9e : SUCCESS in 5m 38s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/a59b524ebe77494cacf6ed7027e96297 : SUCCESS in 6m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f359e8ff3da940a092affcac55ccb51b : SUCCESS in 1h 14m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5ddb9012a79429fad738eb1418293c4 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eccd10c598ba4a3aac00f845f6721861 : SUCCESS in 4m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/61e4aeb7a5a444ad83748b1a7721445a : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82e176107b02448f9fe8955284dbdca7 : SUCCESS in 4m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e54cddb2c3241e1989f9da78ccfed69 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":1},{"id":"b00f377f2a411b6a8c9c1fbb58e7efe04006453c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-21 07:50:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"91021d468c5c99904746ac06a876467dbded5a06","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-21 07:51:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98b71d616f474e70a6a7ab982c971801\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4230061814b34cc884782d6fc3fdb229 : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9bbb242f9664566aa2045961447a77c : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b1bc10dcd017cfcf9dcd60674307beb808ae2d94","revisions":{"b1bc10dcd017cfcf9dcd60674307beb808ae2d94":{"kind":"REWORK","_number":1,"created":"2024-09-16 06:24:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/98/929398/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/98/929398/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/98/929398/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/98/929398/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/98/929398/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/98/929398/1"}}},"commit":{"parents":[{"commit":"a59dba4479f9f391b63901b847d25a77093a460a","subject":"Merge \"Remove [database] use_tpool and TpoolDbapiWrapper\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/a59dba4479f9f391b63901b847d25a77093a460a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-16 06:23:37.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-16 06:23:37.000000000","tz":540},"subject":"Drop SQLALCHEMY_WARN_20","message":"Drop SQLALCHEMY_WARN_20\n\nThis environment was used by SQLAlchemy 1.4 and is no longer necessary\nsince SQLAlchemy was bumped to 2.0 .\n\nChange-Id: I48c1904ddfe27d13733741b5f1ca2661b8f1d86a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/b1bc10dcd017cfcf9dcd60674307beb808ae2d94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/b1bc10dcd017cfcf9dcd60674307beb808ae2d94"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
