)]}'
{"id":"openstack%2Foslo.db~931926","triplet_id":"openstack%2Foslo.db~master~I41bc7b5faccb0012166b64abe8c50ce41037817e","project":"openstack/oslo.db","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-09 15:11:51.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I41bc7b5faccb0012166b64abe8c50ce41037817e","subject":"Declare Python 3.12 support","status":"MERGED","created":"2024-10-09 13:38:38.000000000","updated":"2024-10-10 03:08:15.000000000","submitted":"2024-10-10 03:07:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931926","meta_rev_id":"2e6b2a877d1e642c5551a9218652e9c304362f6a","_number":931926,"virtual_id_number":931926,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-10 03:07:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-10-10 00:59:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-10-10 00:59:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-09 15:10:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d46d8e7c68ef3066c685f93dc91c18c113344fe5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-09 13:38:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"492361c085fee21fbdeff26ab05aab8c4a8a0a32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 15:10:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d127e371c6544ad8b34d6ec9f5b758b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5c35ad0a0a964c40af63c09b82746531 : SUCCESS in 1m 29s\n- oslodb-tox-py38-tips https://zuul.opendev.org/t/openstack/build/6540f8c76bff48dbb60c1737adfaba9c : SUCCESS in 5m 39s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/dbe03e7f1d7c4f47a29fdbef516ce281 : SUCCESS in 4m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2f0c691e9d274551b8a1f8e28e688da4 : FAILURE in 1h 26m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9edeba9ea0248d59e7c4cf04042edb2 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4927cb00bf224ce9b6537110995a78c0 : SUCCESS in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6eb2bcc400e641f4884bd6bae849c74f : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9dfad72c86c410e8ed54d1e9359d97a : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"38d24adf2f58d9e891aaeb99df7d2feb3f49e3ec","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-09 15:11:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6de4163cbe66a3ea1e555e057896b8b9b49643bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 17:16:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d74dc40275bf47358e80ad933bb32aa4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4947ce7a504d4e7d810f704cd5e517f9 : SUCCESS in 1m 37s\n- oslodb-tox-py38-tips https://zuul.opendev.org/t/openstack/build/d47c08bbd5fc4b77a49df9679f071159 : SUCCESS in 5m 01s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/127b5cc1a39946daab25164467181a33 : SUCCESS in 5m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6286f951adaf497080180d0c2c706ce1 : SUCCESS in 1h 38m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b138ff28a48c4e739e75fefae6867bdb : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e47a08ee789a49bbb18344618979e09f : SUCCESS in 4m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/12521615b1be478c81ee54ca2e8e42c4 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf113571987945388f837cc062634377 : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":1},{"id":"57b62651e47d7111fabd9e052ff98926be06136b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-10 00:59:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a443a7c2d3cfd224a9632a574aee594d55e7e6e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 01:00:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ed64827344e8c7fe9dd648866fc25cba2404aadb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 03:07:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a09075df7264498ea7918678b8162e55\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/da4da17318b447109a932048c949b274 : SUCCESS in 1m 33s\n- oslodb-tox-py38-tips https://zuul.opendev.org/t/openstack/build/5b7d0b4c615a45a0bc755a67022ad9e1 : SUCCESS in 4m 58s\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/8ec22859aad44ec1a6071c1d463106a8 : SUCCESS in 4m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/27cfb6294bb0480f86039efbb3cd070b : SUCCESS in 1h 59m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4129329a0c89410391f9ce5f9a56a23b : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3936ede17358401cb43ae2799a49132c : SUCCESS in 3m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/faed15671dab4a51989631e15ca9158d : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d075077827c4c8fb75ae91fa7870b34 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"6762f999feebd80b0dd70d050827e5c6f1082cac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 03:07:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2e6b2a877d1e642c5551a9218652e9c304362f6a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 03:08:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdf9f330f8ce4371acb2e90ee8adf40f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcd7d59a877a48c2b5d4fabc28bea7dc : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"15faddd7a141eebdf99c36e70e6740fe2c56f70a","revisions":{"15faddd7a141eebdf99c36e70e6740fe2c56f70a":{"kind":"REWORK","_number":1,"created":"2024-10-09 13:38:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/931926/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/26/931926/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/26/931926/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/26/931926/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/26/931926/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/26/931926/1"}}},"commit":{"parents":[{"commit":"41ae8dda3983fed005f81834c33370855d653020","subject":"Merge \"Drop SQLALCHEMY_WARN_20\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/41ae8dda3983fed005f81834c33370855d653020"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-09 13:38:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-09 13:38:29.000000000","tz":540},"subject":"Declare Python 3.12 support","message":"Declare Python 3.12 support\n\nNow Python 3.12 unit tests job is voting. So we can safely say this\nis tested with Python 3.12 .\n\nChange-Id: I41bc7b5faccb0012166b64abe8c50ce41037817e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/15faddd7a141eebdf99c36e70e6740fe2c56f70a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/15faddd7a141eebdf99c36e70e6740fe2c56f70a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
