)]}'
{"id":"openstack%2Foslo.db~932460","triplet_id":"openstack%2Foslo.db~master~Ic3a9f6ac78ec9eb12e8a80809dacef7543fb5bc8","project":"openstack/oslo.db","branch":"master","topic":"remove-py38","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-23 14:25:27.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ic3a9f6ac78ec9eb12e8a80809dacef7543fb5bc8","subject":"Remove Python 3.8 support","status":"MERGED","created":"2024-10-16 01:23:07.000000000","updated":"2024-10-23 16:03:03.000000000","submitted":"2024-10-23 16:00:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"932460-remove-py38","meta_rev_id":"9e33d49a3157e871b801679b9cd218e29b48b63e","_number":932460,"virtual_id_number":932460,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-23 16:00:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-10-23 14:25:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2024-10-17 06:44:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-10-23 14:25:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-16 04:30:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-17 06:44:55.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"fac0145a054e86a691989f55b619e2931723efb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-16 01:23:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d102c0fd2440dc41f0d2ff0265f555e5e19ab07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-16 02:40:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7976720becd2a438b0ae0869793f91beef9631e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 04:30:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0956992dfc4943bbb1230c5030329b67\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bca588f33a8346c5b5fa1996bb8943c1 : SUCCESS in 2m 02s\n- oslodb-tox-py39-tips https://zuul.opendev.org/t/openstack/build/041b902b7f1c4e90a33d5916c864442f : SUCCESS in 5m 17s\n- oslodb-tox-py312-tips https://zuul.opendev.org/t/openstack/build/8913c8d828ee45ad8ca4fefabee770c6 : SUCCESS in 6m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/829a3789d5114ce0832a654add339827 : FAILURE in 1h 33m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb25471031ae4bc69301acfe23d304e2 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a882ce0ad1744f88110532af3a8253c : SUCCESS in 4m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3375efb59f0c46cea4a3eb830d2a6041 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb2f5827e59344ffae7eaaaaeb6c3157 : SUCCESS in 3m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e78da551e3b475e83cc6490d179a7ec : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":2},{"id":"fc9c40b5759744d6d708787fdbff180a477197a1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-16 13:39:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e87e222dcedcc162623bbb50580539329c8aad4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 15:00:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d9ee52cbe324101a12e13eb22d76269\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d7c8c054b81f4759b09847cf5653d7f5 : SUCCESS in 1m 44s\n- oslodb-tox-py39-tips https://zuul.opendev.org/t/openstack/build/fd7f6bea98fe4766b6b66b4684fd94bf : SUCCESS in 5m 23s\n- oslodb-tox-py312-tips https://zuul.opendev.org/t/openstack/build/2708a705967445efaf2dd02f51d602e6 : SUCCESS in 7m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/03b9689aab3a48ff94ea5a3efac931de : SUCCESS in 1h 20m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c0aa550033848d784ab40651e14d76b : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0bfe24e21bd4262ba66af6768d53bc2 : SUCCESS in 4m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4674f10e724540a9b3cb7ec44a8e752a : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6586a1cc0824c4b807f0afde08cc798 : SUCCESS in 3m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a59b2487bc404b0cba0359084a722ab7 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":2},{"id":"6813021a8d7750c84d1183f113b86401b165382b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-10-17 06:44:55.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"409d751629361cc8be9a3bd87b144f707220bd5b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-23 14:25:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"66bcc389ed42dfd18fa255d136f488d1e22ffacd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 14:26:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6ba2ca2dd15b398b4e190b471e155b1b75e13072","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 16:00:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c10bdbf913e4a9b9895274f7ad30a5d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/32e946367f1d4f7188b84ecb01ae180e : SUCCESS in 1m 50s\n- oslodb-tox-py39-tips https://zuul.opendev.org/t/openstack/build/862cba11ece54137b8248aa0c7213f69 : SUCCESS in 5m 13s\n- oslodb-tox-py312-tips https://zuul.opendev.org/t/openstack/build/659f2c7be4cd4863b9974b5f75fd188a : SUCCESS in 5m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d57481f090ee462e9762e91216ad27d4 : SUCCESS in 1h 23m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d2bb14c85444f4eb54fb7957ed7af02 : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa9aa76a829f4e0da0e724ecdd9d55b5 : SUCCESS in 4m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d027d4d803da424a896131967634523b : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2147adfe232439892f5f7c05a80a873 : SUCCESS in 4m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6735f1b1ec1e480ea6d7370ae6697462 : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":2},{"id":"4cc766209097cd37483ccb6ac1320742af9d0ca1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 16:00:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9e33d49a3157e871b801679b9cd218e29b48b63e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 16:03:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c860f5eafbed4dbb990e924d56a5ac2e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b19c2d00bc04918b3fb01a9861728a3 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/342fd17dae7e475ab3da4b62cb44e709 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"078a67bb2ef406248536fef46b7cba49e06a0143","revisions":{"5b7344b271ea33599a34406bfcafe1020e9b919a":{"kind":"REWORK","_number":1,"created":"2024-10-16 01:23:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/932460/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/60/932460/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/60/932460/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/60/932460/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/60/932460/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/60/932460/1"}}},"commit":{"parents":[{"commit":"15faddd7a141eebdf99c36e70e6740fe2c56f70a","subject":"Declare Python 3.12 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/15faddd7a141eebdf99c36e70e6740fe2c56f70a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-16 01:22:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-16 01:22:44.000000000","tz":540},"subject":"Remove Python 3.8 support","message":"Remove Python 3.8 support\n\nPython 3.8 is no longer part of the tested runtimes for 2024.2[1]\nbecause its EOL is coming soon.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2024.2.html\n\nChange-Id: Ic3a9f6ac78ec9eb12e8a80809dacef7543fb5bc8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/5b7344b271ea33599a34406bfcafe1020e9b919a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/5b7344b271ea33599a34406bfcafe1020e9b919a"}]},"branch":"refs/heads/master"},"078a67bb2ef406248536fef46b7cba49e06a0143":{"kind":"REWORK","_number":2,"created":"2024-10-16 02:40:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/60/932460/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/60/932460/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/60/932460/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/60/932460/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/60/932460/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/60/932460/2"}}},"commit":{"parents":[{"commit":"15faddd7a141eebdf99c36e70e6740fe2c56f70a","subject":"Declare Python 3.12 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/15faddd7a141eebdf99c36e70e6740fe2c56f70a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-16 01:22:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-16 02:39:45.000000000","tz":540},"subject":"Remove Python 3.8 support","message":"Remove Python 3.8 support\n\nPython 3.8 is no longer part of the tested runtimes for 2024.2[1]\nbecause its EOL is coming soon.\n\nAlso update the python versions in tips job to reflect the supported\nversions.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2024.2.html\n\nChange-Id: Ic3a9f6ac78ec9eb12e8a80809dacef7543fb5bc8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/078a67bb2ef406248536fef46b7cba49e06a0143"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/078a67bb2ef406248536fef46b7cba49e06a0143"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
