)]}'
{"id":"openstack%2Foslo.db~933822","triplet_id":"openstack%2Foslo.db~master~I648727e2525bc873d2385c106b581b54e7a59ede","project":"openstack/oslo.db","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-11-14 12:03:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I648727e2525bc873d2385c106b581b54e7a59ede","subject":"Remove explicit pbr dependency","status":"MERGED","created":"2024-10-31 01:51:35.000000000","updated":"2024-11-14 12:04:29.000000000","submitted":"2024-11-14 12:03:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933822","meta_rev_id":"cdd9a2639d3dec0316e34a35eeea0053136bcf80","_number":933822,"virtual_id_number":933822,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-14 12:03:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-14 10:27:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-11-14 10:27:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-31 02:53:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-14 10:27:25.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"}],"messages":[{"id":"ac5469c5fd10a871255b3bae1b17ad10d39bbe46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-31 01:51:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fccabd85e4367dc35993d9b45a8f0e0d16ebf63f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 02:53:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0eacfbee2d1a4670b191541b6478320f\n\n- oslodb-tox-py39-tips https://zuul.opendev.org/t/openstack/build/04e7d5262f02406da1e17269474b1412 : SUCCESS in 5m 13s\n- oslodb-tox-py312-tips https://zuul.opendev.org/t/openstack/build/6242030989a34d889c93f53c043331af : SUCCESS in 6m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e6aacb38cb8c40a9b88ee6ffef0361aa : SUCCESS in 2m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a31933953ceb45ae976579391310d00f : SUCCESS in 1h 00m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdaaf4f207554df1838db705b1545efd : SUCCESS in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e788b79c53f64c7ea268e15607bafeb0 : SUCCESS in 4m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f2f654f62ae84de6a09256e5385ef5f5 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae12ae99ab4c42698013b38ee0ac1971 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"7ac05f47916cca27cd0a47af643360c3bd17927e","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-11-14 10:27:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c6e4344d2470d8938c03a20ef5147649b8904d8a","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-11-14 10:27:29.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"78ba7a6e58182b50b056f9341c2e65c8b7747c74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 10:28:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e2f09cf9f4a0835736f1393d42f6d8c2ec4714c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 12:03:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b11b936cb1440bb9f319ef95a4fafdb\n\n- oslodb-tox-py39-tips https://zuul.opendev.org/t/openstack/build/b09b890cfd4a46d8a20e0647fd72fcbb : SUCCESS in 5m 47s\n- oslodb-tox-py312-tips https://zuul.opendev.org/t/openstack/build/73928a8c2a354d6cb48b2ccbe59d34bd : SUCCESS in 6m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3c3ce32526fa46159ebfd68e8bd290bd : SUCCESS in 2m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c59b0c03cec74204a7b2eb9860820ac2 : SUCCESS in 1h 35m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b878c2d940c64543abfe241414ee6e1c : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57d0293b70c64f5eb34af95cbfcbb236 : SUCCESS in 3m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/031c0fe1473e4b3a85515566904f55cd : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82317b82980e46a1a54e3dbc77c6c5c7 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":1},{"id":"64b8a777239dbc44cefda7e0079b57eb564fdadb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 12:03:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cdd9a2639d3dec0316e34a35eeea0053136bcf80","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 12:04:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82164f836e604a249d7f72d8f0126ecd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5563d2f89dee41749eb0ddcd281a3ba5 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ef6850a6445890755e91c1d8496163d7a88e8df9","revisions":{"ef6850a6445890755e91c1d8496163d7a88e8df9":{"kind":"REWORK","_number":1,"created":"2024-10-31 01:51:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/933822/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/22/933822/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/22/933822/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/22/933822/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/22/933822/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/22/933822/1"}}},"commit":{"parents":[{"commit":"cf95959442f3af922ce88ca708ffff0d1ab37304","subject":"Add note about requirements lower bounds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/cf95959442f3af922ce88ca708ffff0d1ab37304"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-31 01:41:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-31 01:41:18.000000000","tz":540},"subject":"Remove explicit pbr dependency","message":"Remove explicit pbr dependency\n\npbr is not use at runtime but is used during installation. It should\nbe pulled by setuptools automatically.\n\nChange-Id: I648727e2525bc873d2385c106b581b54e7a59ede\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/ef6850a6445890755e91c1d8496163d7a88e8df9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/ef6850a6445890755e91c1d8496163d7a88e8df9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
