)]}'
{"id":"openstack%2Foslo.db~956138","triplet_id":"openstack%2Foslo.db~master~If8691df33f7ff1f376d5e9fe0f6ddc1ba4de8348","project":"openstack/oslo.db","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-07-31 12:04:58.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2025-07-31 13:15:23.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2025-07-31 13:15:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If8691df33f7ff1f376d5e9fe0f6ddc1ba4de8348","subject":"Uniformize exceptions by removing useless constructors","status":"MERGED","created":"2025-07-30 09:24:08.000000000","updated":"2025-07-31 14:05:15.000000000","submitted":"2025-07-31 13:15:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956138","meta_rev_id":"41976e652ae0a175b770d5517b19441a126f0c45","_number":956138,"virtual_id_number":956138,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-31 13:15:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-07-31 12:04:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2025-07-31 14:05:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-07-31 12:04:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-30 09:47:17.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2025-07-30 09:47:20.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2025-07-30 10:22:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-30 14:38:56.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2025-07-31 14:05:15.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"639b3d15f60d54089d1dfe555c668044787b301c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-07-30 09:24:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac8f1d1c1fb081af9ca5010f128449e7fbd09365","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-30 09:47:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c62f94353bb67d7f4dca576f71364d310078ae74","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-30 09:47:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"807b34b444566605446d2b645f9bd85904188548","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 10:22:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bcbbd3f4fe7481eb440fb12acb777f4\n\n- oslodb-tox-py39-tips https://zuul.opendev.org/t/openstack/build/c6daa3734e43482a9395072f0ebe381e : SUCCESS in 2m 51s\n- oslodb-tox-py312-tips https://zuul.opendev.org/t/openstack/build/8dcf1bc4dec14cb8a08eef88db0998e0 : SUCCESS in 2m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/22e5bb0ca8ce4ec7b7adaa75edad3ede : SUCCESS in 57m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5462d42a930147b28289bf6386491fa0 : SUCCESS in 2m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e016578d2ec8426b8aee245df2adb9dc : SUCCESS in 2m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e212bdce8f9e4a8f952913c8346d708c : SUCCESS in 2m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e93b6640d1f64c72a935b0a3629607fd : SUCCESS in 3m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18518dad27564f82849499b6eb0e0c5c : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":1},{"id":"5ea3bf9a77f929c2eb405b392c97d6b7facfd547","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2025-07-30 14:38:56.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b41d5a6d331d95b8907480a458113344ee29455c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-07-31 08:57:34.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"33cd979e4e6e62c2775eab326d20488e9d6ffc85","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-07-31 08:57:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"180030cff742e26a922593ac57d6cb99960efc0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-31 10:47:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8b1cec98f6b4f058175ee9cd8608bb8\n\n- oslodb-tox-py39-tips https://zuul.opendev.org/t/openstack/build/0d67a7f41b584212a7462e42abae6f66 : SUCCESS in 3m 40s\n- oslodb-tox-py312-tips https://zuul.opendev.org/t/openstack/build/34ced0a6052944f48374a2741d3ce7e5 : SUCCESS in 3m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/430e6df562a54637bda8773e8d00a920 : SUCCESS in 1h 10m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6356db0a08134bd7b9602726d4b9e36c : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df2b0f042aee41e0abb92cf7ea922be6 : SUCCESS in 3m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0b5f0075c4ad4e0592e96209fdf9a1ff : SUCCESS in 3m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/10dd9a7253854cc8960a08656ec6a7bc : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f753a1762b6c48df9731486ab308bd2b : SUCCESS in 2m 20s","accounts_in_message":[],"_revision_number":2},{"id":"ac0242ea165cc7ac7e3fbef3a414e8f649cbb689","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-31 12:04:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2a68b6956914037d6fd2459d935858a92900a64f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-31 12:05:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"43c200b002260a51cd98950c3f012a732fc47432","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-31 13:15:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce795e07f4f442cc8f2f1afcd1cef8b6\n\n- oslodb-tox-py39-tips https://zuul.opendev.org/t/openstack/build/5c9cf3935d934b788bba5cf8516d0e7f : SUCCESS in 4m 21s\n- oslodb-tox-py312-tips https://zuul.opendev.org/t/openstack/build/81aa94dcb9db4ed9abbf3b0c1de436d2 : SUCCESS in 3m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2aa665f52476472c8944e4dfb5466b66 : SUCCESS in 1h 09m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66ae305acd184c609e5af73f20592950 : SUCCESS in 2m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d1679c5f01ac42119a2ec6b71539a3c4 : SUCCESS in 2m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c3ba4eecfbcd4f679b721bcc72b7be36 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3364ff899594648866355685a3f088c : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":2},{"id":"599da0a61565374fbb393fad5f325da532db1e7d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-31 13:15:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e7b7cdeb6f705b99c6e253585f29adfe11c5f3ab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-31 13:16:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9114d365eb644e808c645f3229ba7db8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a25bc8b6ab440958b76731c3fa89c60 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2},{"id":"41976e652ae0a175b770d5517b19441a126f0c45","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2025-07-31 14:05:15.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"aa3c8d8a5216a9eb60fbbfe94ed0654c4a210c23","revisions":{"640afeca916e5e5399ac25c0e1e01f2eade2da69":{"kind":"REWORK","_number":1,"created":"2025-07-30 09:24:08.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/38/956138/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/38/956138/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/38/956138/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/38/956138/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/38/956138/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/38/956138/1"}}},"commit":{"parents":[{"commit":"8eb40b47d569dc7fce6a8c1a0e71d44cf108d7cd","subject":"Merge \"Replace deprecated datetime.datetime.utcnow\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/8eb40b47d569dc7fce6a8c1a0e71d44cf108d7cd"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2025-07-30 09:22:13.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2025-07-30 09:22:33.000000000","tz":120},"subject":"Uniformize exceptions by removing useless constructors","message":"Uniformize exceptions by removing useless constructors\n\nSome exceptions do not needs a constructor. This patch\nremove useless constructors.\n\nChange-Id: If8691df33f7ff1f376d5e9fe0f6ddc1ba4de8348\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/640afeca916e5e5399ac25c0e1e01f2eade2da69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/640afeca916e5e5399ac25c0e1e01f2eade2da69"}]},"branch":"refs/heads/master"},"aa3c8d8a5216a9eb60fbbfe94ed0654c4a210c23":{"kind":"REWORK","_number":2,"created":"2025-07-31 08:57:34.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/38/956138/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/38/956138/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/38/956138/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/38/956138/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/38/956138/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/38/956138/2"}}},"commit":{"parents":[{"commit":"8eb40b47d569dc7fce6a8c1a0e71d44cf108d7cd","subject":"Merge \"Replace deprecated datetime.datetime.utcnow\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/8eb40b47d569dc7fce6a8c1a0e71d44cf108d7cd"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2025-07-30 09:22:13.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2025-07-31 08:57:23.000000000","tz":120},"subject":"Uniformize exceptions by removing useless constructors","message":"Uniformize exceptions by removing useless constructors\n\nSome exceptions do not needs a constructor. This patch\nremove useless constructors.\n\nChange-Id: If8691df33f7ff1f376d5e9fe0f6ddc1ba4de8348\nSigned-off-by: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/aa3c8d8a5216a9eb60fbbfe94ed0654c4a210c23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/aa3c8d8a5216a9eb60fbbfe94ed0654c4a210c23"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
