)]}'
{"id":"openstack%2Foslo.db~959772","triplet_id":"openstack%2Foslo.db~stable%2F2025.2~I6e83137c9a26aaa0450764804da77ec8a1ff0096","project":"openstack/oslo.db","branch":"stable/2025.2","topic":"create-2025.2","hashtags":[],"change_id":"I6e83137c9a26aaa0450764804da77ec8a1ff0096","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.2","status":"MERGED","created":"2025-09-05 12:22:54.000000000","updated":"2025-09-05 13:11:30.000000000","submitted":"2025-09-05 13:08:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"959772-create-2025.2","meta_rev_id":"23afa641df63bbae83ad0d706f496015e4a2c97a","_number":959772,"virtual_id_number":959772,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-05 13:08:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2025-09-05 12:56:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2025-09-05 12:56:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-05 12:42:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-05 12:56:10.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"7d680876e210a09afd60aa09de12ddcca1ad0a57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2025-09-05 12:22:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"63437e2c29536bae2c17741eb479cf32079ad1c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 12:42:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8fddd67a4234075b4e82868f8fc50d0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35b9400dd7894d4fb108077c135fd7ec : SUCCESS in 6m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a204b14982b494a8ea45c20c056940a : SUCCESS in 5m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5db7fe6f010e4bf18a7a91db198ce8f6 : SUCCESS in 3m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e224be1f2ac04e51b2742991beb20b1e : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a25af12e4ec4908b7552733ae25fcc5 : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d8c6c0ff78c438eba359fa17312b46f : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"af04bba64e7e9fb345ff47977ee840932a252ff2","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-09-05 12:56:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3a56b1ca93764fd5f0af71fcceb2b7bd07ef881f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 12:56:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9d3bdf55c081abec905c3bd1d11c9e2f80e19e38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 13:08:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed34cbe36b824f849350112aded271b7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88cfc7a63bf6441fafedc69dbbbee567 : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/34c3e25cd6ef414e8e955c90747b30da : SUCCESS in 4m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/78d97225bd40445e84e33b952e7cf682 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ae4801c4623496e898bf824e6aa7623 : SUCCESS in 7m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/089b07a91b62400dbb757f3bec4e49a4 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":1},{"id":"26e289df66fc306443ed20a0d779fb48a407e685","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 13:08:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"23afa641df63bbae83ad0d706f496015e4a2c97a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 13:11:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82ac26160b6b46ed910e84aecd349d08\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b7f4c3a010a4249be9fa92be25e3eda : SUCCESS in 1m 13s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/779127768d0d4d64abe9b69cd2722563 : SUCCESS in 2m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f38429049b18fc66b6b9844bf6445fc59e52c147","revisions":{"f38429049b18fc66b6b9844bf6445fc59e52c147":{"kind":"REWORK","_number":1,"created":"2025-09-05 12:22:54.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/72/959772/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/72/959772/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/72/959772/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/72/959772/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/72/959772/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/72/959772/1"}}},"commit":{"parents":[{"commit":"073b52e245fc43a99a46e32fa5f98c5a3bc854d1","subject":"Update .gitreview for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/073b52e245fc43a99a46e32fa5f98c5a3bc854d1"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-09-05 12:22:52.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-09-05 12:22:52.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.2","message":"Update TOX_CONSTRAINTS_FILE for stable/2025.2\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/2025.2 branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I6e83137c9a26aaa0450764804da77ec8a1ff0096\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f38429049b18fc66b6b9844bf6445fc59e52c147"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/f38429049b18fc66b6b9844bf6445fc59e52c147"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
