)]}'
{"id":"openstack%2Foslo.db~962509","triplet_id":"openstack%2Foslo.db~master~Id1eaeca5bc2ce62ff926b0404a012960c73f04d2","project":"openstack/oslo.db","branch":"master","topic":"py314","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-06-25 20:05:51.000000000","reason":"Change was submitted"},"11816":{"account":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"last_update":"2026-06-25 20:05:51.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-06-25 20:05:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id1eaeca5bc2ce62ff926b0404a012960c73f04d2","subject":"Python3.14: fix tests for AsyncioFacadeTest","status":"MERGED","created":"2025-09-29 15:24:34.000000000","updated":"2026-06-25 20:06:43.000000000","submitted":"2026-06-25 20:05:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962509-py314","meta_rev_id":"c867ec399830eb32af2efc507bab0aeee9a548c8","_number":962509,"virtual_id_number":962509,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-25 20:05:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-06-25 19:07:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-06-25 19:07:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-29 15:34:24.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2025-09-29 17:12:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-25 19:07:03.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"035023c2a7197739a4af7c8e74290554cd675e21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-09-29 15:24:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09ee9a1d094510f35d0b09cd55338b199ca17d90","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2025-09-29 15:34:24.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6a4acd88e90dd69d6c86ed4b39a48765a79c8d38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 17:12:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee3683337095476c8a432ca33a7f3317\n\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/fd51b10c4adb47a0930ee15dde9995ea : SUCCESS in 5m 15s\n- oslodb-tox-py313-tips https://zuul.opendev.org/t/openstack/build/83559b1db2354636a8033b203a911e32 : SUCCESS in 10m 35s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1ba40b00edee42ccabf6968aefab9473 : SUCCESS in 2m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ce854abe3dc940869489f1e267481d76 : SUCCESS in 1h 44m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b569e88168834b31a78c6e0f081f5f11 : SUCCESS in 2m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/79b4986d3d7e4523b184582dc9cc10d0 : SUCCESS in 3m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2029de6cfd614793a4ba62da4c1c78a8 : SUCCESS in 3m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ebb1455dba22484e8830f1acb1483db6 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68d5532586da40c7846477b8ed536289 : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":1},{"id":"cc4ead9763ad130c3559efb6a3f5e7bb9575d121","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-08 10:30:01.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"94c81f1679df6aacd62a283d470bf0c15716258c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-08 11:55:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66ef4f1684d04ba5917fb2abe52cc805\n\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/ed4fb36ccb884477ab8534a1210b063a : SUCCESS in 20m 03s\n- oslodb-tox-py313-tips https://zuul.opendev.org/t/openstack/build/e4f2d6273e0e4933916493760626431d : SUCCESS in 5m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c2b647182ae941cb8be877601cb37cac : SUCCESS in 1m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dfc3c6c934f548908bc067955b3d89b9 : SUCCESS in 1h 24m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/524fad5d3d8a4779980a33acaa28dd02 : SUCCESS in 2m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1b95fe3526df409b92e57ea264646c00 : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8019ec614d294aabb090da0233964729 : SUCCESS in 3m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/967d934f770d4e02a2e3d02ef5118004 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/822d665f4daf4cf08ef3a3c65c353879 : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":2},{"id":"60239b367f7578db46ad97762d37a22faf9734d3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-25 19:07:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fa0194e5d0fcf8eac951245f00c9dba55b2947cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 19:07:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0a6dd133b2bc1eb4586cf14b7cd02270ebc8a956","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 20:05:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d36fc745e80458f802849324ff5305c\n\n- oslodb-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e9158895c84946a1b869232ccafd2735 : SUCCESS in 3m 06s\n- oslodb-tox-py314-tips https://zuul.opendev.org/t/openstack/build/047b653bd3a749bcb231715a218e582b : SUCCESS in 9m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bc6052ed075548e59ceb47b798301ee2 : SUCCESS in 2m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b066c507160d4a32ac99915e5404032f : SUCCESS in 57m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae3d78b4dbff467dbe01b4ba99715e7d : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1e4a045148b54ceebcaa9537cefd6876 : SUCCESS in 2m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/62477286f1b443dfb6454025464badba : SUCCESS in 3m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0c89deebb052450ebffe649baa1fbbe3 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39ff2be14f3f4d2d96c213498372c973 : SUCCESS in 3m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e041510a980949f7b58b88c2761dda86 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":2},{"id":"860f7c4ba76cad4e08327a5b690fcd8754af933c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 20:05:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c867ec399830eb32af2efc507bab0aeee9a548c8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 20:06:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66b447a4ab42459da953b2227e9c0114\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/495914f7a70841938e37ffaa3ad88b3f : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b1c640dc871dcd74e53d2ae70b8aac35b874ccab","revisions":{"591b9a63a60fae97a1c2009810beafa27ccd6bd1":{"kind":"REWORK","_number":1,"created":"2025-09-29 15:24:34.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/09/962509/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/09/962509/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/09/962509/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/09/962509/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/09/962509/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/09/962509/1"}}},"commit":{"parents":[{"commit":"2170b164bd42d9ea997e7e376c4d361c607cd8c2","subject":"Merge \"Drop redundant library check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/2170b164bd42d9ea997e7e376c4d361c607cd8c2"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-09-29 15:13:58.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-09-29 15:24:28.000000000","tz":120},"subject":"Python3.14: fix tests for AsyncioFacadeTest","message":"Python3.14: fix tests for AsyncioFacadeTest\n\nIn Python3.14, AsyncioFacadeTest tests fail with:\n\n\"ValueError: the greenlet library is required to use this function. No\nmodule named \u0027greenlet\u0027\"\n\nIt is unclear why it works in Python 3.13, but SQLAlchemy does state\nthat one should install SQLAlchemy[asyncio] in order to use greenlet:\n\n\"The SQLAlchemy asyncio module requires that the Python \u0027greenlet\u0027\nlibrary is installed. In order to ensure this dependency is available,\nuse the \u0027sqlalchemy[asyncio]\u0027 install target: \u0027pip install\nsqlalchemy[asyncio]\u0027\"[1]\n\n[1] https://github.com/sqlalchemy/sqlalchemy/blob/2e058575c237f33058da79433f6f1ce5f6739121/lib/sqlalchemy/util/concurrency.py#L40\n\nChange-Id: Id1eaeca5bc2ce62ff926b0404a012960c73f04d2\nSigned-off-by: Cyril Roelandt \u003ccyril@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/591b9a63a60fae97a1c2009810beafa27ccd6bd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/591b9a63a60fae97a1c2009810beafa27ccd6bd1"}]},"branch":"refs/heads/master"},"b1c640dc871dcd74e53d2ae70b8aac35b874ccab":{"kind":"REWORK","_number":2,"created":"2026-03-08 10:30:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/09/962509/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/09/962509/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/09/962509/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/09/962509/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/09/962509/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/09/962509/2"}}},"commit":{"parents":[{"commit":"fc79b296fbaad07177ebb5a8a18896abe8976176","subject":"Delay string interpolations at logging calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/fc79b296fbaad07177ebb5a8a18896abe8976176"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2025-09-29 15:13:58.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-08 10:29:08.000000000","tz":540},"subject":"Python3.14: fix tests for AsyncioFacadeTest","message":"Python3.14: fix tests for AsyncioFacadeTest\n\nIn Python3.14, AsyncioFacadeTest tests fail with:\n\n\"ValueError: the greenlet library is required to use this function. No\nmodule named \u0027greenlet\u0027\"\n\nIt is unclear why it works in Python 3.13, but SQLAlchemy does state\nthat one should install SQLAlchemy[asyncio] in order to use greenlet:\n\n\"The SQLAlchemy asyncio module requires that the Python \u0027greenlet\u0027\nlibrary is installed. In order to ensure this dependency is available,\nuse the \u0027sqlalchemy[asyncio]\u0027 install target: \u0027pip install\nsqlalchemy[asyncio]\u0027\"[1]\n\n[1] https://github.com/sqlalchemy/sqlalchemy/blob/2e058575c237f33058da79433f6f1ce5f6739121/lib/sqlalchemy/util/concurrency.py#L40\n\nChange-Id: Id1eaeca5bc2ce62ff926b0404a012960c73f04d2\nSigned-off-by: Cyril Roelandt \u003ccyril@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/b1c640dc871dcd74e53d2ae70b8aac35b874ccab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/b1c640dc871dcd74e53d2ae70b8aac35b874ccab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
