)]}'
{"id":"openstack%2Foslo.db~962814","triplet_id":"openstack%2Foslo.db~master~I8e2da753dde66f5e6f31a14ac920b933a8acc090","project":"openstack/oslo.db","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-10-08 17:41:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8e2da753dde66f5e6f31a14ac920b933a8acc090","subject":"Migrate bandit options to pyproject.toml","status":"MERGED","created":"2025-10-02 14:44:56.000000000","updated":"2025-10-08 17:42:08.000000000","submitted":"2025-10-08 17:41:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962814","meta_rev_id":"6ccb39e4fa07ca246460b9fc3d659279ff69c6db","_number":962814,"virtual_id_number":962814,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-08 17:41:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-10-08 15:53:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2025-10-02 21:07:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-10-08 15:53:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-02 17:03:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-02 21:07:19.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2025-10-08 15:53:36.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"21dcb7a6fc7a600221ac221f90fbf89e491f4b55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-02 14:44:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2cd4cb3aacef99f4932a87c79cd8ba2186892c18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-02 14:55:25.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8392a922dc76227c5357695dff63bebc6e36405f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 17:03:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c51f67c66cac47b4b31e19f87bff315b\n\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/1262ffd7034e4e1cb02e891cfe2fa05c : SUCCESS in 5m 42s\n- oslodb-tox-py313-tips https://zuul.opendev.org/t/openstack/build/84813da7a2ff4c1bb7446902a2bf39d9 : SUCCESS in 10m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dc744db443f04474ba5431819471cc4b : SUCCESS in 2h 01m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6cb510f689f479d9b1f134bde1abe13 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/86017a0ffca64e409f9c09403df9cf81 : SUCCESS in 4m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/461012a8589b498d81c0acfd14037d4a : SUCCESS in 2m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6a62c1cc76234459a1305f1fcc5ce1e4 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e74639dd63d4d99b440c0cd1864b447 : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":2},{"id":"9c98fe37c9076190993b48ebb6b765e451c3b559","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-10-02 21:07:19.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9cd28396a30b6c7f4264e7abc13299085c71745e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-08 15:53:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"02bbb98fe45d9bfc142653eb01ef9eeb95a0536e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 15:53:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"60bdedaca680d4e013e06f9d887d5eb91903573e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 17:41:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/957d022b21374534b20eb09b73fe1232\n\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/f6c3597327a84c50ac910a669999c70b : SUCCESS in 5m 18s\n- oslodb-tox-py313-tips https://zuul.opendev.org/t/openstack/build/79adcb8fd0ac467890bde6165e5018e0 : SUCCESS in 7m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/35f417b3c5dc412da343103279ec4b9a : SUCCESS in 1h 46m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1295a2eb08e4d2da9e806320a5caafe : SUCCESS in 6m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/85930779d0414367a5317403ec9698f8 : SUCCESS in 5m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bc1618ed76b24251b42613de2aeafd55 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7737a889a4614f52af89b13d7c17d22a : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":2},{"id":"07e5bbb119cd05b98d9485a7647dbb698f5f8b62","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 17:41:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6ccb39e4fa07ca246460b9fc3d659279ff69c6db","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 17:42:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6c44f9d64774852bf76fd814b90ef68\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/718f6bfb88e04c538d1ee5914bb41cdc : SUCCESS in 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e6b87105ee5164966964d87ff476b497fe54e86a","revisions":{"1f440ec13312f892e74a9d4e59dd59ebec6b317c":{"kind":"REWORK","_number":1,"created":"2025-10-02 14:44:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/962814/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/14/962814/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/14/962814/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/14/962814/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/14/962814/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/14/962814/1"}}},"commit":{"parents":[{"commit":"2170b164bd42d9ea997e7e376c4d361c607cd8c2","subject":"Merge \"Drop redundant library check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/2170b164bd42d9ea997e7e376c4d361c607cd8c2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 14:44:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 14:44:47.000000000","tz":540},"subject":"Migrate bandit options to pyproject.toml","message":"Migrate bandit options to pyproject.toml\n\n... so that these options can be used by a different tools than\ntox/pre-commit more easily.\n\nChange-Id: I8e2da753dde66f5e6f31a14ac920b933a8acc090\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/1f440ec13312f892e74a9d4e59dd59ebec6b317c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/1f440ec13312f892e74a9d4e59dd59ebec6b317c"}]},"branch":"refs/heads/master"},"e6b87105ee5164966964d87ff476b497fe54e86a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-10-02 14:55:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/962814/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/14/962814/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/14/962814/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/14/962814/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/14/962814/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/14/962814/2"}}},"commit":{"parents":[{"commit":"2170b164bd42d9ea997e7e376c4d361c607cd8c2","subject":"Merge \"Drop redundant library check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/2170b164bd42d9ea997e7e376c4d361c607cd8c2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 14:44:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-02 14:55:25.000000000","tz":0},"subject":"Migrate bandit options to pyproject.toml","message":"Migrate bandit options to pyproject.toml\n\n... so that these options can be used by different tools than\ntox/pre-commit more easily.\n\nChange-Id: I8e2da753dde66f5e6f31a14ac920b933a8acc090\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e6b87105ee5164966964d87ff476b497fe54e86a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/e6b87105ee5164966964d87ff476b497fe54e86a"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
