)]}'
{"id":"openstack%2Foslo.db~977532","triplet_id":"openstack%2Foslo.db~master~Ie58a3b0a6df7e145570627256fcb82d4cfdfe090","project":"openstack/oslo.db","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-02-27 00:37:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie58a3b0a6df7e145570627256fcb82d4cfdfe090","subject":"Replace usage of undocumented logging.WARN","status":"MERGED","created":"2026-02-21 13:57:58.000000000","updated":"2026-02-27 00:38:55.000000000","submitted":"2026-02-27 00:37:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977532","meta_rev_id":"328e24db635e4e4f7134c1884b77fbbe10e38683","_number":977532,"virtual_id_number":977532,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-27 00:37:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-02-23 11:32:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2026-02-23 07:55:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-02-26 23:04:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-21 19:03:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-23 07:55:03.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2026-02-23 11:32:01.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"bceff9fe086eee652ecf6e3405910e2c7990b391","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-21 13:57:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a8815ad322118135a34b2b8c13fd86ba39acc28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-21 14:54:40.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"c80812a021287842316bdbd017202b4f4080a89c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-21 14:55:37.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"2d5143690e53f56ebc6ffc73324d60b06b92397b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-21 19:03:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/37c7c08bc4c446ec8245d55cb496aade\n\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/a0ce0d5d9b82427382417a1275979c9e : FAILURE in 5m 28s\n- oslodb-tox-py313-tips https://zuul.opendev.org/t/openstack/build/c3d3b869a5734e0ab1b20cd8274715bf : FAILURE in 5m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fce70d31de80434ea94b1f363a58001e : SUCCESS in 1h 38m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15d6d13a099d4e988ba6937c337fcabb : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/acf478035e6246c4b79a9721c954cfc9 : SUCCESS in 2m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e0b0987871c049a5a896b93357fa6687 : SUCCESS in 2m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7b07c689f03447e7b089a3c4b16cf7fb : SUCCESS in 8m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cfdaa27c20a4a34bdfe21359d448c7d : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":3},{"id":"a2f18636573a0e54e9a0acb713dcabb448dedb5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-22 07:05:55.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"1d0c8cfd7dad781fc97ad0bf64227eac3cfa0093","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-22 08:09:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7d40af2836d456794a8841ef2e07d75\n\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/62c25850c8714dd9ad1934d5ec70af79 : SUCCESS in 5m 08s\n- oslodb-tox-py313-tips https://zuul.opendev.org/t/openstack/build/b2b177fff84445d9a479882bda27de3b : SUCCESS in 4m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6a97517df9da4959b540d38842a70140 : SUCCESS in 57m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/090e671deb704f1ba5d6703661333376 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cead91d8af894c39b806c84273feb1a4 : SUCCESS in 2m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f65274600ad847e1b8476cae39dd41d7 : SUCCESS in 3m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/671f38e129914edc9d0e6e92096f3e1c : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a824d7a643b44a7d827dd9738c007cfc : SUCCESS in 2m 27s","accounts_in_message":[],"_revision_number":4},{"id":"c38b89d3eb8e4867e8c159f8dc584241bb0d8831","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-02-23 07:55:03.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"c1f63b556949835e3251078bc2267787823924be","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-23 11:32:01.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"673abd85a9114134a8b8132710da47eead413c7a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-26 23:04:25.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"3d25e71964f6d43df69a5fc6ef72bfaaed8224d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 23:04:50.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ce59acfe618bfbebba9be8de9d566be13217c366","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 00:37:50.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ddebe52f88b4cb19157b5a850d5b0c3\n\n- oslodb-tox-py310-tips https://zuul.opendev.org/t/openstack/build/39b1a704f4164aa4b5de727187cf170f : SUCCESS in 7m 26s\n- oslodb-tox-py313-tips https://zuul.opendev.org/t/openstack/build/a778bc55d31b4423a7aef300b62fe312 : SUCCESS in 6m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/08fcf4d309464b02963b32d7dc51f3ed : SUCCESS in 1h 29m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83bddd7f574d454e9e438cbdf2284cee : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/efcacb5dc1ac41b1a063eba1261f901d : SUCCESS in 3m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b99d0802a2434f4886886263adad98e4 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e03adf2146fa485aa4319d40665c7d9f : SUCCESS in 8m 41s","accounts_in_message":[],"_revision_number":4},{"id":"b1c173780c3957322a26214d088693de78720080","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 00:37:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"328e24db635e4e4f7134c1884b77fbbe10e38683","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 00:38:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e0dbd693aed4688bfb10c0b3b6048a4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e7842fd7860431aac35081f2f638f50 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"18db4152c05d73fb1b6b09787bf04233ed42ad46","revisions":{"ea41d2d4bd9b9e8ef80857c1326f5baeac6ed77d":{"kind":"REWORK","_number":1,"created":"2026-02-21 13:57:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/977532/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/32/977532/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/32/977532/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/32/977532/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/32/977532/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/32/977532/1"}}},"commit":{"parents":[{"commit":"fc79b296fbaad07177ebb5a8a18896abe8976176","subject":"Delay string interpolations at logging calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/fc79b296fbaad07177ebb5a8a18896abe8976176"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-21 13:57:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-21 13:57:09.000000000","tz":540},"subject":"Replace usage of undocumented logging.WARN","message":"Replace usage of undocumented logging.WARN\n\nChange-Id: Ie58a3b0a6df7e145570627256fcb82d4cfdfe090\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/ea41d2d4bd9b9e8ef80857c1326f5baeac6ed77d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/ea41d2d4bd9b9e8ef80857c1326f5baeac6ed77d"}]},"branch":"refs/heads/master"},"bf64ef567d176a8622da14bf98f8a9dba59edc41":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-02-21 14:54:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/977532/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/32/977532/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/32/977532/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/32/977532/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/32/977532/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/32/977532/2"}}},"commit":{"parents":[{"commit":"dea29356941f918d8f354e4cb66d9bd6938596d6","subject":"Fix unit test failure with SQLAlchemy 2.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/dea29356941f918d8f354e4cb66d9bd6938596d6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-21 13:57:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-21 14:54:40.000000000","tz":0},"subject":"Replace usage of undocumented logging.WARN","message":"Replace usage of undocumented logging.WARN\n\nChange-Id: Ie58a3b0a6df7e145570627256fcb82d4cfdfe090\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/bf64ef567d176a8622da14bf98f8a9dba59edc41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/bf64ef567d176a8622da14bf98f8a9dba59edc41"}]},"branch":"refs/heads/master","description":"Rebase"},"bf13329db63b6d6cb5916a533cfa2ff62dfed472":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-02-21 14:55:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/977532/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/32/977532/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/32/977532/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/32/977532/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/32/977532/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/32/977532/3"}}},"commit":{"parents":[{"commit":"d0b4b8167bef95ab983dca2349ae90fb1437acb7","subject":"Fix unit test failure with SQLAlchemy 2.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/d0b4b8167bef95ab983dca2349ae90fb1437acb7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-21 13:57:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-21 14:55:37.000000000","tz":0},"subject":"Replace usage of undocumented logging.WARN","message":"Replace usage of undocumented logging.WARN\n\nChange-Id: Ie58a3b0a6df7e145570627256fcb82d4cfdfe090\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/bf13329db63b6d6cb5916a533cfa2ff62dfed472"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/bf13329db63b6d6cb5916a533cfa2ff62dfed472"}]},"branch":"refs/heads/master","description":"Rebase"},"18db4152c05d73fb1b6b09787bf04233ed42ad46":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-02-22 07:05:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/977532/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/32/977532/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/32/977532/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/32/977532/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/32/977532/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/32/977532/4"}}},"commit":{"parents":[{"commit":"25cd79074ed25876a0fd31dafc09aa84cf77eb2a","subject":"Fix unit test failure with SQLAlchemy 2.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/25cd79074ed25876a0fd31dafc09aa84cf77eb2a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-21 13:57:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-22 07:05:55.000000000","tz":0},"subject":"Replace usage of undocumented logging.WARN","message":"Replace usage of undocumented logging.WARN\n\nChange-Id: Ie58a3b0a6df7e145570627256fcb82d4cfdfe090\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/18db4152c05d73fb1b6b09787bf04233ed42ad46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/18db4152c05d73fb1b6b09787bf04233ed42ad46"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
