)]}'
{"id":"openstack%2Foslo.db~992684","triplet_id":"openstack%2Foslo.db~master~Ia0bb12210e0ec61ed1e18af7c27eb5bb7492888f","project":"openstack/oslo.db","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-06-19 17:21:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia0bb12210e0ec61ed1e18af7c27eb5bb7492888f","subject":"Remove Python 3.10 support","status":"MERGED","created":"2026-06-10 15:06:04.000000000","updated":"2026-06-19 17:22:27.000000000","submitted":"2026-06-19 17:21:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"992684","meta_rev_id":"60c3d7967a0f5ca74fa15434515f4cb84b332e7f","_number":992684,"virtual_id_number":992684,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-19 17:21:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-06-19 15:50:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-06-19 15:50:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-10 17:17:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-19 15:50:54.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"16db97d086fbd7f7ef4f6c490e08238b7d56d654","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-10 15:06:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f7dafd236da17b4c0aa42b6da8ca47cb29d031dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 17:17:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3da2d4acae0143a49c8ad16db543cfae\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/db000585bd5b4160a2543b0fe76a8c0c : SUCCESS in 2m 28s\n- oslodb-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e33ac86557af48e69ec71e3b9012e69a : SUCCESS in 5m 21s\n- oslodb-tox-py314-tips https://zuul.opendev.org/t/openstack/build/fe2bbf826ad54f4393bada1272842433 : SUCCESS in 9m 33s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e8826ea1482b423badbb580ec98c5c4a : SUCCESS in 3m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8a9cee4fb8b04c239b976074ef918e40 : SUCCESS in 2h 04m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3719b6dfbb24a40bda0628e630d69a9 : SUCCESS in 5m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8b7bc089fad145ee88a592857d6077a0 : SUCCESS in 4m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/535b63d1cbff4cfc98c42360cfd1cd9b : SUCCESS in 6m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6c49ac90c9824f5f8639f3ac36c351f2 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cb34a707b724d40bfac4b6053346e55 : SUCCESS in 5m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8d5c13c21cb4fcaa6318ccdc427a959 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"809db3e05e1958639f213f9a3494856c6812ab41","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-19 15:50:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"63ded5031d1a9e34263cbe2513ec8369f29a77e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 15:51:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"16242ccc19efdd8f31ac2a995ceb1ea60c59f216","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 17:21:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce59b09469654e03bf2ef523f0129cfa\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0600628012dd42d880f1e5e7177629ee : SUCCESS in 2m 49s\n- oslodb-tox-py311-tips https://zuul.opendev.org/t/openstack/build/87b30daa939c49bf8706b3df7ce18298 : SUCCESS in 3m 23s\n- oslodb-tox-py314-tips https://zuul.opendev.org/t/openstack/build/c44a9415219c4f39adbddedc28053e78 : SUCCESS in 8m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a577e9c3b75748d195757440210e39ae : SUCCESS in 2m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a07457d9bfe94679b1cd59a52311cc97 : SUCCESS in 1h 28m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3800ecf7ba8544609e6a2533ee1e7fc5 : SUCCESS in 5m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1d32a37f491a416cb0417aaee2c035e6 : SUCCESS in 5m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/abedeb1f7f02441a887d535e3cae1f1d : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e707ecbfd78e4dedb857fc6146beccc6 : SUCCESS in 4m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1dbbddafedc04abfbecdaa25fbaf6b47 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"a65e6aaadbbca53fc7bbe592f3d72359f4b0d20d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 17:21:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"60c3d7967a0f5ca74fa15434515f4cb84b332e7f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-19 17:22:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6020255b0fe545b5b2d230a42d0121dc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7a02244db344bcba6ce4a79c31ee456 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6ec84cf4415479f8ebe87c0609411d7 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8cc1080b01179cfa038d3c95423f305d574d0eee","revisions":{"8cc1080b01179cfa038d3c95423f305d574d0eee":{"kind":"REWORK","_number":1,"created":"2026-06-10 15:06:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/84/992684/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.db","ref":"refs/changes/84/992684/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/84/992684/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/84/992684/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.db refs/changes/84/992684/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.db refs/changes/84/992684/1"}}},"commit":{"parents":[{"commit":"94e926a1a18cfd1b6fdb0b9194f6912a59ac8960","subject":"clean up pyproject.toml deprecated and missing entries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/94e926a1a18cfd1b6fdb0b9194f6912a59ac8960"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-10 15:05:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-06-10 15:05:53.000000000","tz":540},"subject":"Remove Python 3.10 support","message":"Remove Python 3.10 support\n\nPython 3.10 is no longer part of the tested runtimes.\n\nChange-Id: Ia0bb12210e0ec61ed1e18af7c27eb5bb7492888f\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/8cc1080b01179cfa038d3c95423f305d574d0eee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.db/commit/8cc1080b01179cfa038d3c95423f305d574d0eee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
