)]}'
{"id":"openstack%2Foslo.i18n~679321","triplet_id":"openstack%2Foslo.i18n~master~I0c617937f5af7467d1454f72acd0474ae2ce0293","project":"openstack/oslo.i18n","branch":"master","topic":"bug/1841796","hashtags":[],"change_id":"I0c617937f5af7467d1454f72acd0474ae2ce0293","subject":"Deprecate Message.translate in favor of Message.translation","status":"MERGED","created":"2019-08-29 19:16:47.000000000","updated":"2019-09-23 15:15:36.000000000","submitted":"2019-09-23 15:12:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679321-1569251569347-dadc74e4","meta_rev_id":"b6e813b7e89d41744e08e16b2aba580f0f9d9f83","_number":679321,"virtual_id_number":679321,"owner":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-23 15:12:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2019-09-23 12:01:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-09-06 09:43:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2019-09-23 12:01:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-06 09:43:45.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-09-23 12:01:12.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-09-23 15:12:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6b4251a7b8b23d8e6ae562d507b09e588a6a1088","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-08-29 19:16:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"57e90ca3c44f76bebdf3050002271aa99a769cf2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 21:03:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/dfb2a230365e4ff8b3f8d3a12379e7ec : SUCCESS in 1h 38m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c71d3a0b1e004414ad6d4cce96cbad72 : SUCCESS in 1h 36m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c961dbe0e25c4077a5ab305a370de479 : SUCCESS in 3m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a09069615c3040dab1b58f23951b95fa : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b8dafdeec3d41c9b042ab12a00fc72d : SUCCESS in 3m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a74d1b31e0f24ef79f54ae6916023787 : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32717c0501c444e4abdc56884ec88307 : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/67f7ed630d804254bb5213a149838cab : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebd633b977a4451ba39f2729ca6a536d : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":1},{"id":"7fa0eae1024f0a81f3f0c81c9698de9e0d75cc13","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-09-06 09:43:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3b0d13e5304ce75182ef0fa1a4b685754e9516ab","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-09-23 12:01:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f2517a4cb3e775608eb4447a2b7bf1fb43cdc86f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 12:02:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f5680766261bffe2a0a9e31f9a78252baeb10bae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 13:10:56.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"28184bcc144f174114a1c932d1eff56b716ede15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 15:12:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e1d49dc38a95bc16e47457764c34095a3f9a01ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 15:12:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/877bf61319cc4f8a9316ddc593304a21 : SUCCESS in 1h 34m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5c8cd92fd3dc40d0b424669473fb1e1d : SUCCESS in 1h 41m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b564b71f29de468799e9c1653b4b0964 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7bba1d308eb4c18a8ffdaed3c26367a : SUCCESS in 4m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/acddb96e277f43639be37b8c6a74e706 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/160240dc22624e7d9b3a82407e3f8555 : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/de4ba918f70c4bfbb6d3a84b78609ec5 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55f144aa23054e5e9a2b08c461ba4f9d : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":1},{"id":"64ac18973b565638c39f870262aff4ac5e5c6594","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 15:15:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1148dd4a87f34634a588ac013eb4f4fd : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"22499258465e4606e234f8aad99e56338b3e0722","revisions":{"22499258465e4606e234f8aad99e56338b3e0722":{"kind":"REWORK","_number":1,"created":"2019-08-29 19:16:47.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/21/679321/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.i18n","ref":"refs/changes/21/679321/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.i18n refs/changes/21/679321/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.i18n refs/changes/21/679321/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.i18n refs/changes/21/679321/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.i18n refs/changes/21/679321/1"}}},"commit":{"parents":[{"commit":"f8fa059c46fed84514004ece7dccacf6380cf263","subject":"Allow Message.translate to handle unhashable inputs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.i18n/commit/f8fa059c46fed84514004ece7dccacf6380cf263"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-08-29 19:10:42.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-08-29 19:10:42.000000000","tz":0},"subject":"Deprecate Message.translate in favor of Message.translation","message":"Deprecate Message.translate in favor of Message.translation\n\nIt shadows the unicode function of the same name, and there\u0027s no\nway to entirely disambiguate the two based on the parameter passed\nin. This change deprecates Message.translate and makes it a wrapper\naround the new function, Message.translation.\n\nNote that we never documented calling Message.translate directly.\nThe documented way to get the translated form of a Message is to call\nthe _translate.translate function in this project, so chances are\nthat this public API change will have little actual impact on users.\n\nChange-Id: I0c617937f5af7467d1454f72acd0474ae2ce0293\nCloses-Bug: 1841796\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.i18n/commit/22499258465e4606e234f8aad99e56338b3e0722"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.i18n/commit/22499258465e4606e234f8aad99e56338b3e0722"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
