)]}'
{"id":"openstack%2Foslo.limit~1008652","triplet_id":"openstack%2Foslo.limit~master~I42536dbd1856cc857196ef5205d866997e89fb5d","project":"openstack/oslo.limit","branch":"master","hashtags":[],"change_id":"I42536dbd1856cc857196ef5205d866997e89fb5d","subject":"Reuse test commands in cover target","status":"MERGED","created":"2026-10-04 03:28:42.000000000","updated":"2026-10-04 07:14:15.000000000","submitted":"2026-10-04 07:13:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1008652","meta_rev_id":"b1d926725b6a6b7e22a796b920072dd033d1307d","_number":1008652,"virtual_id_number":1008652,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-10-04 07:13:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-10-04 05:40:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-10-04 05:40:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-10-04 05:25:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a5bcb658f7b489cb23e2e958c95165860e26aae5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-10-04 03:28:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab8d2bd9d84336750099543eb1f10a3b5417acd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-04 05:25:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18599b23c2c743a6a69ee8902e45dd4d\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b16daa7a16c94a0482ef7418da313621 : SUCCESS in 1m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/277f393505f44b92822db1d7563b02e7 : SUCCESS in 2m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fc4903843af34de7be180c0f04528e4b : SUCCESS in 1h 43m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb9ad642e4ec4f7289ac2f356ee9c8a2 : SUCCESS in 3m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ba87c0879f27462788ad5b9bf0d12a04 : SUCCESS in 2m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/446c93eb8b124010885e58532bf51137 : SUCCESS in 3m 28s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/0852c738ca39496e90997457e0ba65fa : SUCCESS in 8m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/566202327576410d8bfb0b79cb394408 : SUCCESS in 4m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e224d486e0d7496689b2a101e38f0dec : SUCCESS in 2m 53s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/b06cec04019c41409064171ac13eece6 : SUCCESS in 11m 07s","accounts_in_message":[],"_revision_number":1},{"id":"3a506f813ce8d7505e2bc682f99a5cacaeb3ca91","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-10-04 05:40:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b4bdbe897f13da2d8230996e7977871d32b6f61d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-04 05:40:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fb25673f7cd00e3ea52c6f5161b2801f563bd2b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-04 07:13:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5163f1a6ecf74e3e80cd8b0e4b0dc7c4\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5ab90393e8224ea78a5c666581f298d5 : SUCCESS in 4m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2e9f14250ec04127985091329026be9f : SUCCESS in 1h 26m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/712955969a7b4758b44945dee31e79ea : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5008e4435de54441b0c4df4bfec21602 : SUCCESS in 2m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/932aa1b279c64cc28de91e78542c0d97 : SUCCESS in 3m 13s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/620705236d984bb5b664356f7a606cf6 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fd56a4bd7c84fc7a7c6df05624b55d9 : SUCCESS in 3m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/573f6cb24a4e42bd99b0e7c2b4cc4e12 : SUCCESS in 4m 33s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/5ff345a511844a61938b65be693fc2b5 : SUCCESS in 26m 42s","accounts_in_message":[],"_revision_number":1},{"id":"2fcdde8114d36a277f56aba35fb7d223bb777372","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-04 07:13:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b1d926725b6a6b7e22a796b920072dd033d1307d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-04 07:14:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c858a93f19e4f3dae3f0d37e0ba2539\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b4d2e853e624eec9e1721d2279a7008 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7b88369315f427d85363da14ae0fcf3 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2167be5926e8a7318018c8c38eb5faa2be89fbd1","revisions":{"2167be5926e8a7318018c8c38eb5faa2be89fbd1":{"kind":"REWORK","_number":1,"created":"2026-10-04 03:28:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/1008652/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/52/1008652/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/52/1008652/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/52/1008652/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/52/1008652/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/52/1008652/1"}}},"commit":{"parents":[{"commit":"3752e4bcfa38f5d54eb3b21e0854be31a004bc26","subject":"Use pyproject.toml to maintain coverage options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/3752e4bcfa38f5d54eb3b21e0854be31a004bc26"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-10-04 03:27:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-10-04 03:27:51.000000000","tz":540},"subject":"Reuse test commands in cover target","message":"Reuse test commands in cover target\n\n... to maintain the commands in a single place.\n\nAlso install coverage only when needed.\n\nChange-Id: I42536dbd1856cc857196ef5205d866997e89fb5d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/2167be5926e8a7318018c8c38eb5faa2be89fbd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/2167be5926e8a7318018c8c38eb5faa2be89fbd1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
