)]}'
{"id":"openstack%2Foslo.limit~654730","triplet_id":"openstack%2Foslo.limit~master~Ibb5a0f72dd8e7dab4c42340f26226fdc6451d523","project":"openstack/oslo.limit","branch":"master","hashtags":[],"change_id":"Ibb5a0f72dd8e7dab4c42340f26226fdc6451d523","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"MERGED","created":"2019-04-23 05:25:34.000000000","updated":"2019-04-26 23:31:54.000000000","submitted":"2019-04-26 23:31:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654730-1556321514231-47422c5d","meta_rev_id":"8fe09413521c83b7abf4b63b9998867d2d0edeaa","_number":654730,"virtual_id_number":654730,"owner":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":2,"date":"2019-04-26 23:31:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":2,"date":"2019-04-26 18:44:31.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-04-26 21:41:05.000000000","_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-04-26 21:41:05.000000000","_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-23 06:29:00.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2019-04-26 18:44:31.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-04-26 21:41:05.000000000","updated_by":{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},"reviewer":{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},"state":"REVIEWER"},{"updated":"2019-04-26 23:31:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aeb6e37b50485a4ca46ec0f2aead1814d0d6cd29","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-04-23 05:25:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ffb3c02321102f23f95f6a19f970ecbc9203a99","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-04-23 06:29:00.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7b522d2f9bef143480055c233ce3ed885deb0118","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-04-23 06:30:36.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0e7d63a6a716b2a461ea373c01e7782fa52a7538","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 09:05:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/30/654730/2/check/test-release-openstack/df713fc/ : SUCCESS in 3m 41s\n- tempest-full http://logs.openstack.org/30/654730/2/check/tempest-full/91cd156/ : SUCCESS in 1h 42m 19s\n- tempest-full-py3 http://logs.openstack.org/30/654730/2/check/tempest-full-py3/e1aa796/ : SUCCESS in 1h 56m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/30/654730/2/check/openstack-tox-lower-constraints/74d58d1/ : SUCCESS in 4m 39s\n- openstack-tox-pep8 http://logs.openstack.org/30/654730/2/check/openstack-tox-pep8/0ccc452/ : SUCCESS in 4m 37s\n- openstack-tox-py27 http://logs.openstack.org/30/654730/2/check/openstack-tox-py27/698cf97/ : SUCCESS in 3m 53s\n- openstack-tox-py36 http://logs.openstack.org/30/654730/2/check/openstack-tox-py36/a8e1eb2/ : SUCCESS in 4m 26s\n- openstack-tox-py37 http://logs.openstack.org/30/654730/2/check/openstack-tox-py37/b85e037/ : SUCCESS in 4m 41s\n- openstack-tox-docs http://logs.openstack.org/30/654730/2/check/openstack-tox-docs/9222460/html/ : SUCCESS in 4m 23s\n- build-openstack-releasenotes http://logs.openstack.org/30/654730/2/check/build-openstack-releasenotes/80bba18/html/ : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":2},{"id":"dfe673a74f1561e93592cb2ba9e75dbb975583aa","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-04-26 18:44:31.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6f500a9d3cc96ac5fa2664bd5add1cf393eabbab","author":{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},"date":"2019-04-26 21:41:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"550dad6a20be8541db450ae8296f962835893fa6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 21:41:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7fe081740ffde2b158cb73f1bd9b30eb2a824860","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 23:31:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/30/654730/2/gate/test-release-openstack/c7be7ee/ : SUCCESS in 5m 25s\n- tempest-full http://logs.openstack.org/30/654730/2/gate/tempest-full/557ba36/ : SUCCESS in 1h 45m 29s\n- tempest-full-py3 http://logs.openstack.org/30/654730/2/gate/tempest-full-py3/3829772/ : SUCCESS in 1h 28m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/30/654730/2/gate/openstack-tox-lower-constraints/fa2223a/ : SUCCESS in 4m 25s\n- openstack-tox-pep8 http://logs.openstack.org/30/654730/2/gate/openstack-tox-pep8/cbe8bf8/ : SUCCESS in 4m 57s\n- openstack-tox-py27 http://logs.openstack.org/30/654730/2/gate/openstack-tox-py27/50838f6/ : SUCCESS in 4m 50s\n- openstack-tox-py36 http://logs.openstack.org/30/654730/2/gate/openstack-tox-py36/f46986d/ : SUCCESS in 5m 01s\n- openstack-tox-py37 http://logs.openstack.org/30/654730/2/gate/openstack-tox-py37/b4eb179/ : SUCCESS in 4m 46s\n- openstack-tox-docs http://logs.openstack.org/30/654730/2/gate/openstack-tox-docs/2b5356d/html/ : SUCCESS in 4m 49s\n- build-openstack-releasenotes http://logs.openstack.org/30/654730/2/gate/build-openstack-releasenotes/4468512/html/ : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":2},{"id":"8fe09413521c83b7abf4b63b9998867d2d0edeaa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 23:31:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0b26714e842606c9d3084ffffc31cee2984b7656","revisions":{"f47425f71c4bb31b68edbe256fb8fa87f5263e12":{"kind":"REWORK","_number":1,"created":"2019-04-23 05:25:34.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/30/654730/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/30/654730/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/30/654730/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/30/654730/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/30/654730/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/30/654730/1"}}},"commit":{"parents":[{"commit":"74a10f73280f4f4bd5c1d8f655a36c8bdd1e4f44","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/74a10f73280f4f4bd5c1d8f655a36c8bdd1e4f44"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-04-23 05:25:13.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-04-23 05:25:30.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Ibb5a0f72dd8e7dab4c42340f26226fdc6451d523\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/f47425f71c4bb31b68edbe256fb8fa87f5263e12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/f47425f71c4bb31b68edbe256fb8fa87f5263e12"}]},"branch":"refs/heads/master"},"0b26714e842606c9d3084ffffc31cee2984b7656":{"kind":"REWORK","_number":2,"created":"2019-04-23 06:30:36.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/30/654730/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/30/654730/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/30/654730/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/30/654730/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/30/654730/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/30/654730/2"}}},"commit":{"parents":[{"commit":"74a10f73280f4f4bd5c1d8f655a36c8bdd1e4f44","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/74a10f73280f4f4bd5c1d8f655a36c8bdd1e4f44"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-04-23 05:25:13.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-04-23 06:30:33.000000000","tz":0},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Ibb5a0f72dd8e7dab4c42340f26226fdc6451d523\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/0b26714e842606c9d3084ffffc31cee2984b7656"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/0b26714e842606c9d3084ffffc31cee2984b7656"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
