)]}'
{"id":"openstack%2Foslo.limit~665709","triplet_id":"openstack%2Foslo.limit~master~I50f52189862bc3d109654e158cba381b156e70b6","project":"openstack/oslo.limit","branch":"master","hashtags":[],"change_id":"I50f52189862bc3d109654e158cba381b156e70b6","subject":"Remove verification functionality","status":"MERGED","created":"2019-06-17 15:18:33.000000000","updated":"2019-06-25 18:12:14.000000000","submitted":"2019-06-25 18:12:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665709-1561486334204-3286e172","meta_rev_id":"0424a742b0a7f2824a20fbbc1def014ec5460efa","_number":665709,"virtual_id_number":665709,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"date":"2019-06-20 13:44:43.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":2,"date":"2019-06-25 18:12:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-06-25 16:12:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2019-06-21 21:23:30.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2019-06-25 16:12:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-21 21:23:30.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-06-25 16:12:05.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-06-25 18:12:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6d28e381af2c46b2212cc732e28f03791d079cff","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-06-17 15:18:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c7b12187f6e18e3bfd3890aaf784e7c850d7426","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 17:20:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/09/665709/1/check/tempest-full/a53fa3b/ : SUCCESS in 1h 50m 53s\n- tempest-full-py3 http://logs.openstack.org/09/665709/1/check/tempest-full-py3/e8f6ffd/ : SUCCESS in 1h 41m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/09/665709/1/check/openstack-tox-lower-constraints/fe57590/ : SUCCESS in 4m 35s\n- openstack-tox-pep8 http://logs.openstack.org/09/665709/1/check/openstack-tox-pep8/6b972b6/ : SUCCESS in 4m 08s\n- openstack-tox-py27 http://logs.openstack.org/09/665709/1/check/openstack-tox-py27/9559767/ : SUCCESS in 5m 47s\n- openstack-tox-py36 http://logs.openstack.org/09/665709/1/check/openstack-tox-py36/ac2069f/ : SUCCESS in 4m 33s\n- openstack-tox-py37 http://logs.openstack.org/09/665709/1/check/openstack-tox-py37/7477991/ : SUCCESS in 5m 34s\n- openstack-tox-docs http://logs.openstack.org/09/665709/1/check/openstack-tox-docs/cb87c66/html/ : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":1},{"id":"e61f12c0e7e1c494d2b1f3f9cc89ce95ecfad2af","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-06-19 20:02:11.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"520335e2ff025108438a77f3dd42f058ea5f4d38","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 20:49:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/09/665709/2/check/tempest-full/5fbd380/ : FAILURE in 11m 26s\n- tempest-full-py3 http://logs.openstack.org/09/665709/2/check/tempest-full-py3/61d6def/ : FAILURE in 10m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/09/665709/2/check/openstack-tox-lower-constraints/f9f1a11/ : SUCCESS in 5m 05s\n- openstack-tox-pep8 http://logs.openstack.org/09/665709/2/check/openstack-tox-pep8/488eb35/ : SUCCESS in 4m 13s\n- openstack-tox-py27 http://logs.openstack.org/09/665709/2/check/openstack-tox-py27/f9fcfbe/ : SUCCESS in 3m 58s\n- openstack-tox-py36 http://logs.openstack.org/09/665709/2/check/openstack-tox-py36/30304ea/ : SUCCESS in 6m 54s\n- openstack-tox-py37 http://logs.openstack.org/09/665709/2/check/openstack-tox-py37/074eef7/ : SUCCESS in 6m 39s\n- openstack-tox-docs http://logs.openstack.org/09/665709/2/check/openstack-tox-docs/3651921/html/ : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":2},{"id":"9ba24759bff76c41b67628decef2b66033a4cbe3","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-06-20 13:44:43.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"030c07c9dbef6f272f12bbf924a63b0877cdfc7b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 15:28:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/09/665709/2/check/tempest-full/4ecfef2/ : SUCCESS in 1h 33m 49s\n- tempest-full-py3 http://logs.openstack.org/09/665709/2/check/tempest-full-py3/c26aada/ : SUCCESS in 1h 37m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/09/665709/2/check/openstack-tox-lower-constraints/ed5b706/ : SUCCESS in 5m 51s\n- openstack-tox-pep8 http://logs.openstack.org/09/665709/2/check/openstack-tox-pep8/00f06b7/ : SUCCESS in 4m 22s\n- openstack-tox-py27 http://logs.openstack.org/09/665709/2/check/openstack-tox-py27/deeef68/ : SUCCESS in 5m 07s\n- openstack-tox-py36 http://logs.openstack.org/09/665709/2/check/openstack-tox-py36/9293045/ : SUCCESS in 4m 28s\n- openstack-tox-py37 http://logs.openstack.org/09/665709/2/check/openstack-tox-py37/c498afd/ : SUCCESS in 4m 36s\n- openstack-tox-docs http://logs.openstack.org/09/665709/2/check/openstack-tox-docs/b3a1b78/html/ : SUCCESS in 9m 59s","accounts_in_message":[],"_revision_number":2},{"id":"e1955a1a4203c0937f44d1b4a2830ba47913a903","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-06-21 21:23:30.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1783efa7f9a30401ae44d9aa9cb4ba7f684ad457","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-06-25 16:12:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a8c16a87a18346240f033be4b68e0823cbd1d85a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 16:20:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dfb9fa8904545f3c0887e7fa54a938ff9e4b9f46","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 18:12:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/09/665709/2/gate/tempest-full/4e0b4b8/ : SUCCESS in 1h 43m 39s\n- tempest-full-py3 http://logs.openstack.org/09/665709/2/gate/tempest-full-py3/ec62ed1/ : SUCCESS in 1h 33m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/09/665709/2/gate/openstack-tox-lower-constraints/9da5613/ : SUCCESS in 2m 55s\n- openstack-tox-pep8 http://logs.openstack.org/09/665709/2/gate/openstack-tox-pep8/9d354f3/ : SUCCESS in 2m 48s\n- openstack-tox-py27 http://logs.openstack.org/09/665709/2/gate/openstack-tox-py27/1e4931b/ : SUCCESS in 3m 01s\n- openstack-tox-py36 http://logs.openstack.org/09/665709/2/gate/openstack-tox-py36/7ddbced/ : SUCCESS in 2m 44s\n- openstack-tox-py37 http://logs.openstack.org/09/665709/2/gate/openstack-tox-py37/0a878ae/ : SUCCESS in 3m 39s\n- openstack-tox-docs http://logs.openstack.org/09/665709/2/gate/openstack-tox-docs/0ca09df/html/ : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":2},{"id":"0424a742b0a7f2824a20fbbc1def014ec5460efa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 18:12:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8bf64bb278ba17264d427ac0e07646c4e496a3b0","revisions":{"a222d4f6295d28fdb4c32b07e3d79b85698bf1d7":{"kind":"REWORK","_number":1,"created":"2019-06-17 15:18:33.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/09/665709/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/09/665709/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/09/665709/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/09/665709/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/09/665709/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/09/665709/1"}}},"commit":{"parents":[{"commit":"adcf84a681d2868455ade05bef2c644e93078210","subject":"Remove ProjectClaim object from oslo.limit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/adcf84a681d2868455ade05bef2c644e93078210"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-06-17 14:38:57.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-06-17 14:38:57.000000000","tz":0},"subject":"Remove verification functionality","message":"Remove verification functionality\n\nWe want the initial implementation of oslo.limit to be as simple as\npossible. Originally, we thought we were going to implement resource\nverification, but that\u0027s something we can do later.\n\nResource verification is meant to protect against race conditions when\ntwo clients claim the same amount of resources and the project is over\nlimit. Unfortunately, having services implement resource verification\nrequires them to use either a context manager or make multiple calls\nto enforce limits and sprinkling those calls around existing business\nlogic. Since we don\u0027t know which approach we want to use to implement\nthis, let\u0027s just focus on enforcing limits and come back to resource\nverification when we get more feedback from developers incorporating\nthis into their services.\n\nChange-Id: I50f52189862bc3d109654e158cba381b156e70b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/a222d4f6295d28fdb4c32b07e3d79b85698bf1d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/a222d4f6295d28fdb4c32b07e3d79b85698bf1d7"}]},"branch":"refs/heads/master"},"8bf64bb278ba17264d427ac0e07646c4e496a3b0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-06-19 20:02:11.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/09/665709/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/09/665709/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/09/665709/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/09/665709/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/09/665709/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/09/665709/2"}}},"commit":{"parents":[{"commit":"af184185fe03a6f73c44eedeb6bcf53082a96a64","subject":"Remove ProjectClaim object from oslo.limit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/af184185fe03a6f73c44eedeb6bcf53082a96a64"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-06-17 14:38:57.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-06-19 18:18:19.000000000","tz":0},"subject":"Remove verification functionality","message":"Remove verification functionality\n\nWe want the initial implementation of oslo.limit to be as simple as\npossible. Originally, we thought we were going to implement resource\nverification, but that\u0027s something we can do later.\n\nResource verification is meant to protect against race conditions when\ntwo clients claim the same amount of resources and the project is over\nlimit. Unfortunately, having services implement resource verification\nrequires them to use either a context manager or make multiple calls\nto enforce limits and sprinkling those calls around existing business\nlogic. Since we don\u0027t know which approach we want to use to implement\nthis, let\u0027s just focus on enforcing limits and come back to resource\nverification when we get more feedback from developers incorporating\nthis into their services.\n\nChange-Id: I50f52189862bc3d109654e158cba381b156e70b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/8bf64bb278ba17264d427ac0e07646c4e496a3b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/8bf64bb278ba17264d427ac0e07646c4e496a3b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
