)]}'
{"id":"openstack%2Foslo.limit~666085","triplet_id":"openstack%2Foslo.limit~master~Iae45df091e98904fd6a30db9923b141279fb2cc0","project":"openstack/oslo.limit","branch":"master","topic":"rewrite","hashtags":[],"change_id":"Iae45df091e98904fd6a30db9923b141279fb2cc0","subject":"Add ksa connection logic","status":"MERGED","created":"2019-06-18 17:53:53.000000000","updated":"2019-12-02 17:44:51.000000000","submitted":"2019-12-02 17:43:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666085-1575308596727-a8d7bdcf","meta_rev_id":"611b227671c2cfaa3ef8b1bee009aa009e601ba6","_number":666085,"virtual_id_number":666085,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-02 17:43:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-02 10:08:26.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-12-02 15:54:16.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":2,"date":"2019-11-27 19:20:06.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2019-12-02 15:54:16.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"date":"2019-11-22 16:59:18.000000000","_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-21 13:56:23.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-11-22 16:59:18.000000000","updated_by":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"reviewer":{"_account_id":9954,"name":"Lance Bragstad","username":"lbragstad","inactive":true},"state":"REVIEWER"},{"updated":"2019-11-27 19:20:06.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-12-02 10:08:26.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2019-12-02 15:54:16.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-12-02 17:43:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0592b250e21d72a30b087f82214079197ae8dcc4","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-06-18 17:53:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f43d0d29d6f0995faac1a41ccfc149c84cfad90","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 20:13:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/85/666085/1/check/requirements-check/807f008/ : SUCCESS in 4m 21s\n- tempest-full http://logs.openstack.org/85/666085/1/check/tempest-full/386d2d7/ : TIMED_OUT in 2h 07m 41s\n- tempest-full-py3 http://logs.openstack.org/85/666085/1/check/tempest-full-py3/2fe025c/ : SUCCESS in 1h 30m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/666085/1/check/openstack-tox-lower-constraints/4113c26/ : FAILURE in 4m 38s\n- openstack-tox-pep8 http://logs.openstack.org/85/666085/1/check/openstack-tox-pep8/8d400ec/ : SUCCESS in 4m 14s\n- openstack-tox-py27 http://logs.openstack.org/85/666085/1/check/openstack-tox-py27/a2f256f/ : FAILURE in 3m 48s\n- openstack-tox-py36 http://logs.openstack.org/85/666085/1/check/openstack-tox-py36/8f91bdf/ : FAILURE in 9m 24s\n- openstack-tox-py37 http://logs.openstack.org/85/666085/1/check/openstack-tox-py37/7190578/ : FAILURE in 4m 37s\n- openstack-tox-docs http://logs.openstack.org/85/666085/1/check/openstack-tox-docs/ae2becf/ : FAILURE in 6m 14s","accounts_in_message":[],"_revision_number":1},{"id":"c2cf2ee5cbd688e5df2cff47e33af56cd0822878","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-06-19 20:02:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e07239def90e986f7d5d04cf1984cc3720e2d3d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 22:09:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/85/666085/2/check/tempest-full/4d2f607/ : SUCCESS in 1h 48m 26s\n- tempest-full-py3 http://logs.openstack.org/85/666085/2/check/tempest-full-py3/8c78f72/ : SUCCESS in 1h 42m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/666085/2/check/openstack-tox-lower-constraints/d8d9dc1/ : FAILURE in 7m 37s\n- openstack-tox-pep8 http://logs.openstack.org/85/666085/2/check/openstack-tox-pep8/30e6f4f/ : SUCCESS in 4m 51s\n- openstack-tox-py27 http://logs.openstack.org/85/666085/2/check/openstack-tox-py27/340fa45/ : FAILURE in 4m 34s\n- openstack-tox-py36 http://logs.openstack.org/85/666085/2/check/openstack-tox-py36/453355e/ : FAILURE in 4m 49s\n- openstack-tox-py37 http://logs.openstack.org/85/666085/2/check/openstack-tox-py37/ffdd49c/ : FAILURE in 4m 49s\n- openstack-tox-docs http://logs.openstack.org/85/666085/2/check/openstack-tox-docs/2fe564d/ : FAILURE in 4m 45s","accounts_in_message":[],"_revision_number":2},{"id":"d846dbb25c4c06d49d61a2a1cb4ec98786cebcb6","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-06-20 11:26:30.000000000","message":"Patch Set 2:\n\nmakes sense, although clearly Zuul disagrees about some of the details.","accounts_in_message":[],"_revision_number":2},{"id":"150c7b138c183f03d27a6e0228374b1ce689c811","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-06-20 14:23:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"086b11b24eee3e45946f0ac58a519ef30ba6ddb6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 16:26:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/85/666085/3/check/tempest-full/99d7e09/ : SUCCESS in 1h 51m 21s\n- tempest-full-py3 http://logs.openstack.org/85/666085/3/check/tempest-full-py3/08203fe/ : SUCCESS in 1h 36m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/666085/3/check/openstack-tox-lower-constraints/d192723/ : FAILURE in 4m 38s\n- openstack-tox-pep8 http://logs.openstack.org/85/666085/3/check/openstack-tox-pep8/be30bd7/ : SUCCESS in 5m 06s\n- openstack-tox-py27 http://logs.openstack.org/85/666085/3/check/openstack-tox-py27/21c02fe/ : FAILURE in 9m 53s\n- openstack-tox-py36 http://logs.openstack.org/85/666085/3/check/openstack-tox-py36/98219be/ : FAILURE in 8m 08s\n- openstack-tox-py37 http://logs.openstack.org/85/666085/3/check/openstack-tox-py37/0252547/ : FAILURE in 5m 23s\n- openstack-tox-docs http://logs.openstack.org/85/666085/3/check/openstack-tox-docs/da39f84/ : FAILURE in 5m 06s","accounts_in_message":[],"_revision_number":3},{"id":"a21535f44a6abe153062e8a091923eb5244ed704","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-06-24 15:45:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f65f1cf0dfdab2b02c2ee23bcfbf0a09592b3eca","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-06-24 16:01:34.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"bd2bd1908aa2c75e64045aac5ba9f6aceb8643af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 18:19:32.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/85/666085/5/check/requirements-check/a4c4586/ : FAILURE in 3m 40s\n- tempest-full http://logs.openstack.org/85/666085/5/check/tempest-full/fc4f10d/ : SUCCESS in 2h 06m 17s\n- tempest-full-py3 http://logs.openstack.org/85/666085/5/check/tempest-full-py3/8b59742/ : SUCCESS in 1h 29m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/666085/5/check/openstack-tox-lower-constraints/908a8f8/ : SUCCESS in 4m 28s\n- openstack-tox-pep8 http://logs.openstack.org/85/666085/5/check/openstack-tox-pep8/943eb4d/ : SUCCESS in 4m 15s\n- openstack-tox-py27 http://logs.openstack.org/85/666085/5/check/openstack-tox-py27/e94d929/ : SUCCESS in 4m 38s\n- openstack-tox-py36 http://logs.openstack.org/85/666085/5/check/openstack-tox-py36/f6e1371/ : SUCCESS in 4m 44s\n- openstack-tox-py37 http://logs.openstack.org/85/666085/5/check/openstack-tox-py37/f19b761/ : SUCCESS in 7m 35s\n- openstack-tox-docs http://logs.openstack.org/85/666085/5/check/openstack-tox-docs/d14f6c3/html/ : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":5},{"id":"ae52bcd56216a25b3b5005e161da8151f89143b7","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-06-25 18:53:46.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"3cd7e670fc9bcc909cb5354977e11bd463db9c3d","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-06-25 19:24:05.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"43873bcc4fbeb22785a4e71a15cedb8def8e7e9e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 21:34:48.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/85/666085/7/check/requirements-check/7cb6964/ : SUCCESS in 2m 01s\n- tempest-full http://logs.openstack.org/85/666085/7/check/tempest-full/f66fb49/ : SUCCESS in 2h 03m 45s\n- tempest-full-py3 http://logs.openstack.org/85/666085/7/check/tempest-full-py3/c4be4f8/ : SUCCESS in 1h 35m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/666085/7/check/openstack-tox-lower-constraints/1934e9e/ : SUCCESS in 3m 04s\n- openstack-tox-pep8 http://logs.openstack.org/85/666085/7/check/openstack-tox-pep8/ff20b01/ : SUCCESS in 3m 08s\n- openstack-tox-py27 http://logs.openstack.org/85/666085/7/check/openstack-tox-py27/32dc0b3/ : SUCCESS in 3m 19s\n- openstack-tox-py36 http://logs.openstack.org/85/666085/7/check/openstack-tox-py36/7456880/ : SUCCESS in 3m 12s\n- openstack-tox-py37 http://logs.openstack.org/85/666085/7/check/openstack-tox-py37/27865a3/ : SUCCESS in 3m 35s\n- openstack-tox-docs http://logs.openstack.org/85/666085/7/check/openstack-tox-docs/089742d/html/ : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":7},{"id":"91d9e5868b4fa6e815ebadb0fe9a4b6d291266d4","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-06-26 15:54:16.000000000","message":"Topic set to rewrite","accounts_in_message":[],"_revision_number":7},{"id":"609155788d314d1bb7d4a7bf285d1999a10d75f4","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-06-27 22:51:41.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"5a09474df94fd6b717b466236cd88c273e42e5d4","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-07-01 16:25:14.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"c6d6ec5d6a3bda3f9a0f450d7e0579b78f1cf739","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-07-01 16:44:58.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"37456f5a4c963e5def486cf5774f86662a128992","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-07-01 16:45:48.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"08a1b2f17b5437fb0c658cbba9cd8af10cddcaf0","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-07-01 16:47:16.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"20d9f1f6662a526f8578ac580d2872f6ebb492b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 18:50:36.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/85/666085/9/check/requirements-check/44c4543/ : SUCCESS in 2m 18s\n- tempest-full http://logs.openstack.org/85/666085/9/check/tempest-full/eb480bc/ : SUCCESS in 1h 56m 21s\n- tempest-full-py3 http://logs.openstack.org/85/666085/9/check/tempest-full-py3/867609f/ : SUCCESS in 1h 42m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/666085/9/check/openstack-tox-lower-constraints/cd6e2f0/ : SUCCESS in 3m 17s\n- openstack-tox-pep8 http://logs.openstack.org/85/666085/9/check/openstack-tox-pep8/e8c13ee/ : SUCCESS in 4m 28s\n- openstack-tox-py27 http://logs.openstack.org/85/666085/9/check/openstack-tox-py27/93ad3aa/ : SUCCESS in 3m 15s\n- openstack-tox-py36 http://logs.openstack.org/85/666085/9/check/openstack-tox-py36/21e8116/ : SUCCESS in 3m 08s\n- openstack-tox-py37 http://logs.openstack.org/85/666085/9/check/openstack-tox-py37/37dfc4d/ : SUCCESS in 3m 17s\n- openstack-tox-docs http://logs.openstack.org/85/666085/9/check/openstack-tox-docs/7ad0aa7/html/ : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":9},{"id":"4f321c6cae4e51092109488baae02b97f89ff381","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-07-01 20:33:27.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"381455685c60c3a951ff0aa48c00404615818a9b","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-07-02 16:54:30.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"72c64d10a4f40b3b272c7cc1ad97d67b20f8d46f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-08-21 13:56:23.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"30a772cf4da04d1f5be9023d3a19b72aa49ec8f9","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-11-22 14:38:03.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"edade7fa3687d9318e50c4c1413e3ae70894677a","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-11-22 14:48:12.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"c26b3933fdc4b0ef5c9049d0372d4495fbe8492a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 16:47:26.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/eafa6b249f8e4a21a0b91ed11b8ad27a : SUCCESS in 2m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8c6e84607046480cb3bb16a6fdaa92a5 : SUCCESS in 1h 52m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6a23f99152bb4c20bed0380bbbc76d3a : SUCCESS in 1h 23m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/047fb40505d64c759acfe503d4df0482 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7206bef62374a96997378bf5ebf327f : SUCCESS in 4m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/787de00dfd894649b808aa33ad1f9506 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/875e0acc5a7b4351b513857c214b1f2f : SUCCESS in 3m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7eb6c1533cf545ccb23ccd1f66e4d9e5 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a733c11126e546e38610aad3368df820 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":10},{"id":"84d09a33d77456322a25d63662d977febec79e3f","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-11-22 16:59:18.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"5b7f3354859578aeae5295c1507bb2076275333f","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-11-22 18:16:30.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"42ade7e35d4d49b1a2086167c2664b6e921a0ee9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 18:54:03.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a9ead0bfe887478a8a5e9d6434d3e685 : SUCCESS in 2m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/64a5119418954ef5b13eb55e373b66d2 : SUCCESS in 1h 31m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fd7e5de21b5b4ce6936c9a3763a12c4e : FAILURE in 1h 43m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04a9b704e3cd41f988a41e3076f8ba88 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f64d00c4ce545f3875adeb59a03573d : SUCCESS in 3m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0d1dfdf3aefd4c6583f5fd91c034f1ae : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac44e25922884f8780d9e3098c4de361 : SUCCESS in 3m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c6fff5e5ec8d424b8f93f73e784ef1c6 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e7ca1b7af2f49fca2941b120ae92332 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":11},{"id":"84020b262da6630b0ab1142e0ee43a0ba60f4a98","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-11-26 21:04:29.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"cc3555b0b71768d10666c1e90600d54b0c07cf64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 22:38:33.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2b467af9f1f44a299e560a1d762540ff : SUCCESS in 1m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/177f2e35a1074cd4a5f6731377f975cb : SUCCESS in 1h 29m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1e9954eb7f4d4ef7b4d06c934df92fec : SUCCESS in 1h 13m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66be942c33784dc79513dca52015ebdf : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d33f7219a084289a3feeb712a13d984 : SUCCESS in 2m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/af821b765b524aada5d81acfe1e61ead : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/279c0e62768e4f94a1a5d5fdd331c4aa : FAILURE in 14m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5e071276d32a47da9fde21ec40648d87 : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73689e5cb0da431caa71bdc1a14eac2f : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":11},{"id":"4217f94d87ac0c96290d44428548551203f9270f","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-11-27 16:55:50.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"f0ca8faa90f6c3490a6c4f6a79be76de3c0b19af","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-11-27 17:41:11.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"d5851059ed3669509f4c3d77e780aa99e4e4359b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 18:55:28.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/08e156f0e37f4389b29b0e948acfd5cf : SUCCESS in 2m 48s\n- tempest-full https://zuul.opendev.org/t/openstack/build/083fedf1f1314cdbb498255cbc0d578d : SUCCESS in 1h 51m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5fab4bab19734e4e8937a3fb20c661a8 : SUCCESS in 1h 35m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62640dc3ddc8446d965b0071a5277466 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7ff6d2008c34073969e4ac7ea4c502b : SUCCESS in 2m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/84fa93591ddb4cb59a73c756438961b3 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f546467eaa04f32afd003d5c2b5c823 : SUCCESS in 3m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2a585660f70d40178d4c77430a456261 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdaf904a8c054c929cd4cd9dfaea690e : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":11},{"id":"5d5550629cde3886d2de451661de3b49ca6d5e88","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-11-27 19:20:06.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"2c0d4e647df4d0fe52d5c04841c5fb6c49a8c573","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-11-27 20:05:35.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"63433a5c6fb4aad2ca42080825cc61f7fbb09cd8","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-12-02 10:08:26.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"cd025e5a0e5d4201af1dae420a007d6fe6ee928c","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-12-02 15:54:16.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\n(1 comment)\n\nLet\u0027s do this!","accounts_in_message":[],"_revision_number":11},{"id":"78906b4007b9d470699156e33091fe83ef645f2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 15:54:27.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"0e2f17bde49025c713d5c4632dbda563e0ee5220","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 17:43:16.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f673eb5187b34b9fadb054429945387b : SUCCESS in 4m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6dbcac5d80794974a17b5099f8e8bc94 : SUCCESS in 1h 43m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6b47c89b4da5406796d7bbb2519d1c77 : SUCCESS in 1h 34m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7061f419dcf740e5969f179ae7d9f451 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95d34ecedb3d424f9547e5881db426ff : SUCCESS in 4m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8de8cf8f4c5a483983c514375a19606d : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3978b49789a48a8ba8206ada98080be : SUCCESS in 6m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/26b8ea732d5348c880b4ec6835c20330 : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f79097a15774a6e84f57da1b4504d65 : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":11},{"id":"b3306173f8a63415e0c3e84652324b4e28a61b95","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 17:43:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"611b227671c2cfaa3ef8b1bee009aa009e601ba6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 17:44:51.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec68a516420c454d84dad87f17749c29 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"c02c025a66f1ad64a9626d9cefb962c51d5eedb7","revisions":{"6df34c20c25b35b0bb5c456ff11e939765e75d44":{"kind":"REWORK","_number":1,"created":"2019-06-18 17:53:53.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/85/666085/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/85/666085/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/1"}}},"commit":{"parents":[{"commit":"614a923e089195ce0206fe4bb903d0c0d138fae5","subject":"Add skeleton enforce() method to Enforcer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/614a923e089195ce0206fe4bb903d0c0d138fae5"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-06-18 17:52:29.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-06-18 17:52:29.000000000","tz":0},"subject":"WIP: Add ksa connection logic","message":"WIP: Add ksa connection logic\n\nChange-Id: Iae45df091e98904fd6a30db9923b141279fb2cc0\nCo-Authored-By: wangxiyuan \u003cwangxiyuan1007@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/6df34c20c25b35b0bb5c456ff11e939765e75d44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/6df34c20c25b35b0bb5c456ff11e939765e75d44"}]},"branch":"refs/heads/master"},"bb08648554ccffa1f6f3ab04571933bbe51c682c":{"kind":"REWORK","_number":2,"created":"2019-06-19 20:02:11.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/85/666085/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/85/666085/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/2"}}},"commit":{"parents":[{"commit":"4bbd0cba0949cdad104ea4d3f809f520d13913ad","subject":"Add skeleton enforce() method to Enforcer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/4bbd0cba0949cdad104ea4d3f809f520d13913ad"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-06-18 17:52:29.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-06-19 20:00:30.000000000","tz":0},"subject":"Add ksa connection logic","message":"Add ksa connection logic\n\nThis commit pulls the ksa-specific logic from a previous attempt:\n\n  Ibc35360dd4a952bbe63ee68ddafd5def0608949d\n\nSubsequent patches will use this connection to query unified limit\ninformation out of keystone.\n\nChange-Id: Iae45df091e98904fd6a30db9923b141279fb2cc0\nCo-Authored-By: wangxiyuan \u003cwangxiyuan1007@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/bb08648554ccffa1f6f3ab04571933bbe51c682c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/bb08648554ccffa1f6f3ab04571933bbe51c682c"}]},"branch":"refs/heads/master"},"04aadaa94f02557296e3834dfccbfb6e86ed6f4b":{"kind":"REWORK","_number":3,"created":"2019-06-20 14:23:17.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/85/666085/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/85/666085/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/3"}}},"commit":{"parents":[{"commit":"996edbae1293f77282952426b66ffa0bba871397","subject":"Add skeleton enforce() method to Enforcer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/996edbae1293f77282952426b66ffa0bba871397"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-06-18 17:52:29.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-06-20 14:18:40.000000000","tz":0},"subject":"Add ksa connection logic","message":"Add ksa connection logic\n\nThis commit pulls the ksa-specific logic from a previous attempt:\n\n  Ibc35360dd4a952bbe63ee68ddafd5def0608949d\n\nSubsequent patches will use this connection to query unified limit\ninformation out of keystone.\n\nChange-Id: Iae45df091e98904fd6a30db9923b141279fb2cc0\nCo-Authored-By: wangxiyuan \u003cwangxiyuan1007@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/04aadaa94f02557296e3834dfccbfb6e86ed6f4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/04aadaa94f02557296e3834dfccbfb6e86ed6f4b"}]},"branch":"refs/heads/master"},"77da40bbb345a2f8d0c248ea7daa40101100abdf":{"kind":"REWORK","_number":4,"created":"2019-06-24 15:45:44.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/85/666085/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/85/666085/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/4"}}},"commit":{"parents":[{"commit":"996edbae1293f77282952426b66ffa0bba871397","subject":"Add skeleton enforce() method to Enforcer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/996edbae1293f77282952426b66ffa0bba871397"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-06-18 17:52:29.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-06-24 15:45:35.000000000","tz":0},"subject":"Add ksa connection logic","message":"Add ksa connection logic\n\nThis commit pulls the ksa-specific logic from a previous attempt:\n\n  Ibc35360dd4a952bbe63ee68ddafd5def0608949d\n\nSubsequent patches will use this connection to query unified limit\ninformation out of keystone.\n\nChange-Id: Iae45df091e98904fd6a30db9923b141279fb2cc0\nCo-Authored-By: wangxiyuan \u003cwangxiyuan1007@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/77da40bbb345a2f8d0c248ea7daa40101100abdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/77da40bbb345a2f8d0c248ea7daa40101100abdf"}]},"branch":"refs/heads/master"},"7daf2ad9e75e562252d3319ce031aee3eb99c09b":{"kind":"REWORK","_number":5,"created":"2019-06-24 16:01:34.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/85/666085/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/85/666085/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/5"}}},"commit":{"parents":[{"commit":"996edbae1293f77282952426b66ffa0bba871397","subject":"Add skeleton enforce() method to Enforcer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/996edbae1293f77282952426b66ffa0bba871397"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-06-18 17:52:29.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-06-24 16:01:27.000000000","tz":0},"subject":"Add ksa connection logic","message":"Add ksa connection logic\n\nThis commit pulls the ksa-specific logic from a previous attempt:\n\n  Ibc35360dd4a952bbe63ee68ddafd5def0608949d\n\nSubsequent patches will use this connection to query unified limit\ninformation out of keystone.\n\nChange-Id: Iae45df091e98904fd6a30db9923b141279fb2cc0\nCo-Authored-By: wangxiyuan \u003cwangxiyuan1007@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/7daf2ad9e75e562252d3319ce031aee3eb99c09b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/7daf2ad9e75e562252d3319ce031aee3eb99c09b"}]},"branch":"refs/heads/master"},"07aa7ea99ac893dc9babe4e2db50d300492f10db":{"kind":"REWORK","_number":6,"created":"2019-06-25 18:53:46.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/85/666085/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/85/666085/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/6"}}},"commit":{"parents":[{"commit":"996edbae1293f77282952426b66ffa0bba871397","subject":"Add skeleton enforce() method to Enforcer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/996edbae1293f77282952426b66ffa0bba871397"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-06-18 17:52:29.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-06-25 18:53:36.000000000","tz":0},"subject":"Add ksa connection logic","message":"Add ksa connection logic\n\nThis commit pulls the ksa-specific logic from a previous attempt:\n\n  Ibc35360dd4a952bbe63ee68ddafd5def0608949d\n\nSubsequent patches will use this connection to query unified limit\ninformation out of keystone.\n\nChange-Id: Iae45df091e98904fd6a30db9923b141279fb2cc0\nCo-Authored-By: wangxiyuan \u003cwangxiyuan1007@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/07aa7ea99ac893dc9babe4e2db50d300492f10db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/07aa7ea99ac893dc9babe4e2db50d300492f10db"}]},"branch":"refs/heads/master"},"07f3e2045ce9be046817406cda240681ecfde143":{"kind":"REWORK","_number":7,"created":"2019-06-25 19:24:05.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/85/666085/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/85/666085/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/7"}}},"commit":{"parents":[{"commit":"996edbae1293f77282952426b66ffa0bba871397","subject":"Add skeleton enforce() method to Enforcer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/996edbae1293f77282952426b66ffa0bba871397"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-06-18 17:52:29.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-06-25 19:21:46.000000000","tz":0},"subject":"Add ksa connection logic","message":"Add ksa connection logic\n\nThis commit pulls the ksa-specific logic from a previous attempt:\n\n  Ibc35360dd4a952bbe63ee68ddafd5def0608949d\n\nSubsequent patches will use this connection to query unified limit\ninformation out of keystone.\n\nChange-Id: Iae45df091e98904fd6a30db9923b141279fb2cc0\nCo-Authored-By: wangxiyuan \u003cwangxiyuan1007@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/07f3e2045ce9be046817406cda240681ecfde143"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/07f3e2045ce9be046817406cda240681ecfde143"}]},"branch":"refs/heads/master"},"ea2307dd3f93580d5e957e7359521476dd98e065":{"kind":"REWORK","_number":8,"created":"2019-07-01 16:44:58.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/85/666085/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/85/666085/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/8"}}},"commit":{"parents":[{"commit":"6d0a6d4d7ad6eacf1c954a9958df61bf3bff6104","subject":"Rename filter_resource resource_filters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/6d0a6d4d7ad6eacf1c954a9958df61bf3bff6104"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-06-18 17:52:29.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-07-01 16:29:27.000000000","tz":0},"subject":"Add ksa connection logic","message":"Add ksa connection logic\n\nThis commit pulls the ksa-specific logic from a previous attempt:\n\n  Ibc35360dd4a952bbe63ee68ddafd5def0608949d\n\nSubsequent patches will use this connection to query unified limit\ninformation out of keystone.\n\nChange-Id: Iae45df091e98904fd6a30db9923b141279fb2cc0\nCo-Authored-By: wangxiyuan \u003cwangxiyuan1007@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/ea2307dd3f93580d5e957e7359521476dd98e065"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/ea2307dd3f93580d5e957e7359521476dd98e065"}]},"branch":"refs/heads/master"},"52e2a22cd86bcd338b25d9bc9c24e8dd7d35f013":{"kind":"NO_CHANGE","_number":9,"created":"2019-07-01 16:45:48.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/85/666085/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/85/666085/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/9"}}},"commit":{"parents":[{"commit":"90acc274f7f944e4eab4de2a114253ea24c5e6af","subject":"Rename filter_resource resource_filters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/90acc274f7f944e4eab4de2a114253ea24c5e6af"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-06-18 17:52:29.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-07-01 16:45:35.000000000","tz":0},"subject":"Add ksa connection logic","message":"Add ksa connection logic\n\nThis commit pulls the ksa-specific logic from a previous attempt:\n\n  Ibc35360dd4a952bbe63ee68ddafd5def0608949d\n\nSubsequent patches will use this connection to query unified limit\ninformation out of keystone.\n\nChange-Id: Iae45df091e98904fd6a30db9923b141279fb2cc0\nCo-Authored-By: wangxiyuan \u003cwangxiyuan1007@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/52e2a22cd86bcd338b25d9bc9c24e8dd7d35f013"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/52e2a22cd86bcd338b25d9bc9c24e8dd7d35f013"}]},"branch":"refs/heads/master"},"a5256cebf367fa162d67390ea6b331da6043b439":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-11-22 14:48:12.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/85/666085/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/85/666085/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/10"}}},"commit":{"parents":[{"commit":"91ca7b0e26afdf1dbf1a0e7ff298317e78e7a873","subject":"tox: Keeping going with docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/91ca7b0e26afdf1dbf1a0e7ff298317e78e7a873"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-06-18 17:52:29.000000000","tz":0},"committer":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2019-11-22 14:48:12.000000000","tz":0},"subject":"Add ksa connection logic","message":"Add ksa connection logic\n\nThis commit pulls the ksa-specific logic from a previous attempt:\n\n  Ibc35360dd4a952bbe63ee68ddafd5def0608949d\n\nSubsequent patches will use this connection to query unified limit\ninformation out of keystone.\n\nChange-Id: Iae45df091e98904fd6a30db9923b141279fb2cc0\nCo-Authored-By: wangxiyuan \u003cwangxiyuan1007@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/a5256cebf367fa162d67390ea6b331da6043b439"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/a5256cebf367fa162d67390ea6b331da6043b439"}]},"branch":"refs/heads/master"},"c02c025a66f1ad64a9626d9cefb962c51d5eedb7":{"kind":"NO_CHANGE","_number":11,"created":"2019-11-22 16:59:18.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/85/666085/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/85/666085/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/85/666085/11"}}},"commit":{"parents":[{"commit":"91ca7b0e26afdf1dbf1a0e7ff298317e78e7a873","subject":"tox: Keeping going with docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/91ca7b0e26afdf1dbf1a0e7ff298317e78e7a873"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-06-18 17:52:29.000000000","tz":0},"committer":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2019-11-22 16:58:36.000000000","tz":0},"subject":"Add ksa connection logic","message":"Add ksa connection logic\n\nThis commit pulls the ksa-specific logic from a previous attempt:\n\n  Ibc35360dd4a952bbe63ee68ddafd5def0608949d\n\nSubsequent patches will use this connection to query unified limit\ninformation out of keystone.\n\nChange-Id: Iae45df091e98904fd6a30db9923b141279fb2cc0\nCo-Authored-By: wangxiyuan \u003cwangxiyuan1007@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/c02c025a66f1ad64a9626d9cefb962c51d5eedb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/c02c025a66f1ad64a9626d9cefb962c51d5eedb7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
