)]}'
{"id":"openstack%2Foslo.limit~667452","triplet_id":"openstack%2Foslo.limit~master~I3913267e472caba42ee8f724ee30d46bcdd88cf5","project":"openstack/oslo.limit","branch":"master","topic":"rewrite","hashtags":[],"change_id":"I3913267e472caba42ee8f724ee30d46bcdd88cf5","subject":"Implement flat enforcement model","status":"ABANDONED","created":"2019-06-25 19:54:32.000000000","updated":"2020-05-14 13:44:31.000000000","total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d78d33ca5dc9ea4e8ba7bf0173ec58f93a38c6b3","_number":667452,"virtual_id_number":667452,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-07-03 20:08:11.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":1,"date":"2019-07-01 19:07:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-11-25 12:35:44.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":-1,"date":"2019-07-01 16:47:31.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-27 22:52:11.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-07-01 19:07:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-03 20:08:11.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-11-25 12:35:44.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"}],"messages":[{"id":"226d9e91e3e90324d7f7f805803ff083f5455090","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-06-25 19:54:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bee9c298b51e2c27e9651e4cf3af3233227ad339","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 22:10:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/52/667452/1/check/tempest-full/6636497/ : SUCCESS in 1h 47m 26s\n- tempest-full-py3 http://logs.openstack.org/52/667452/1/check/tempest-full-py3/066c59c/ : SUCCESS in 1h 53m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/52/667452/1/check/openstack-tox-lower-constraints/33af56f/ : SUCCESS in 3m 29s\n- openstack-tox-pep8 http://logs.openstack.org/52/667452/1/check/openstack-tox-pep8/90bf242/ : SUCCESS in 2m 43s\n- openstack-tox-py27 http://logs.openstack.org/52/667452/1/check/openstack-tox-py27/b0a2412/ : SUCCESS in 3m 09s\n- openstack-tox-py36 http://logs.openstack.org/52/667452/1/check/openstack-tox-py36/6b046bf/ : SUCCESS in 3m 04s\n- openstack-tox-py37 http://logs.openstack.org/52/667452/1/check/openstack-tox-py37/bd16a0b/ : SUCCESS in 3m 43s\n- openstack-tox-docs http://logs.openstack.org/52/667452/1/check/openstack-tox-docs/c217e51/html/ : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":1},{"id":"b303af10465c281f34c1cd5b2baa1b96ea3fa6fe","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-06-26 15:54:28.000000000","message":"Topic set to rewrite","accounts_in_message":[],"_revision_number":1},{"id":"9078d42e832d0ef4be80107e373f406e8988749e","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-06-27 22:52:11.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"6049a61115496e6a0d5e1345f9b85142f4dcc1be","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-07-01 16:44:32.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c3b7303e9d417d76916cf057cdb54765c50162d6","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-07-01 16:44:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"44227c587fb68f398b7d4554beb3a0205cbe3730","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-07-01 16:45:48.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"dd4f51486213ada3e9a323b4854d6f4f9bae5e3e","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-07-01 16:47:31.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"085639581fdbab0e9b16bdb49dfbfd1b231ddd94","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 19:07:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/52/667452/3/check/tempest-full/7e429e1/ : SUCCESS in 1h 39m 32s\n- tempest-full-py3 http://logs.openstack.org/52/667452/3/check/tempest-full-py3/66d4a61/ : SUCCESS in 1h 33m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/52/667452/3/check/openstack-tox-lower-constraints/552b71e/ : SUCCESS in 3m 04s\n- openstack-tox-pep8 http://logs.openstack.org/52/667452/3/check/openstack-tox-pep8/cae0e48/ : SUCCESS in 3m 15s\n- openstack-tox-py27 http://logs.openstack.org/52/667452/3/check/openstack-tox-py27/f54a42e/ : SUCCESS in 3m 27s\n- openstack-tox-py36 http://logs.openstack.org/52/667452/3/check/openstack-tox-py36/0ee4515/ : SUCCESS in 3m 49s\n- openstack-tox-py37 http://logs.openstack.org/52/667452/3/check/openstack-tox-py37/9652f72/ : SUCCESS in 4m 10s\n- openstack-tox-docs http://logs.openstack.org/52/667452/3/check/openstack-tox-docs/0c03ee5/html/ : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":3},{"id":"85f494e1475d570d7b879ec509ed26aa15642a96","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-07-02 17:16:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e836229936ba22a566ff405981669ad6c3783073","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-07-03 20:08:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cbf8421a25e224d134d1fc76cd5dd81bb04f2b1a","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-11-25 12:35:44.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nI have reworked this chain into this one: https://review.opendev.org/#/c/695310\n\nI am yet to look at creating a functional test like you have in this change, which would be awesome. I am tempted to say it should be me creating that, so we get someone else trying to write \"client\" code against this interface. Thoughts?","accounts_in_message":[],"_revision_number":3},{"id":"d78d33ca5dc9ea4e8ba7bf0173ec58f93a38c6b3","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-05-14 13:44:31.000000000","message":"Abandoned\n\nI\u0027m going to abandon this since I don\u0027t have the cycles to work on it. If someone else would like to pick it up and run with it they are move than welcome to do so.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f5f66d120acb65f0a66c741534a285b1e75745b5","revisions":{"56dec2a62a60722bd66a6c4e96a924b08e98204c":{"kind":"REWORK","_number":1,"created":"2019-06-25 19:54:32.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/52/667452/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/52/667452/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/52/667452/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/52/667452/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/52/667452/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/52/667452/1"}}},"commit":{"parents":[{"commit":"559ddf17ae70a908602abb16a18a65da19bb394f","subject":"Flush out basic enforcer and model relationship","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/559ddf17ae70a908602abb16a18a65da19bb394f"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-06-25 19:52:03.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-06-25 19:52:03.000000000","tz":0},"subject":"Implement flat enforcement model","message":"Implement flat enforcement model\n\nThis commit attempts to implement the flat enforcement model, where\nproject hierarchy is not taken into account during usage and limit\ncalculation and enforcement.\n\nThis commit also adds some helper methods to _BaseEnforcer to help\ncollect important service and region information.\n\nChange-Id: I3913267e472caba42ee8f724ee30d46bcdd88cf5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/56dec2a62a60722bd66a6c4e96a924b08e98204c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/56dec2a62a60722bd66a6c4e96a924b08e98204c"}]},"branch":"refs/heads/master"},"e407e1749b7b987885d2d9a453b9bf09cfc22d91":{"kind":"REWORK","_number":2,"created":"2019-07-01 16:44:58.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/52/667452/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/52/667452/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/52/667452/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/52/667452/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/52/667452/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/52/667452/2"}}},"commit":{"parents":[{"commit":"3f73dd6a84cbe26211809cc31232756b6ebefd3a","subject":"Flush out basic enforcer and model relationship","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/3f73dd6a84cbe26211809cc31232756b6ebefd3a"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-06-25 19:52:03.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-07-01 16:42:25.000000000","tz":0},"subject":"Implement flat enforcement model","message":"Implement flat enforcement model\n\nThis commit attempts to implement the flat enforcement model, where\nproject hierarchy is not taken into account during usage and limit\ncalculation and enforcement.\n\nThis commit also adds some helper methods to _BaseEnforcer to help\ncollect important service and region information.\n\nChange-Id: I3913267e472caba42ee8f724ee30d46bcdd88cf5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/e407e1749b7b987885d2d9a453b9bf09cfc22d91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/e407e1749b7b987885d2d9a453b9bf09cfc22d91"}]},"branch":"refs/heads/master"},"f5f66d120acb65f0a66c741534a285b1e75745b5":{"kind":"NO_CHANGE","_number":3,"created":"2019-07-01 16:45:48.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/52/667452/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/52/667452/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/52/667452/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/52/667452/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/52/667452/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/52/667452/3"}}},"commit":{"parents":[{"commit":"0613e8154193e5e29c7170dc519af61f44156e7e","subject":"Flush out basic enforcer and model relationship","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/0613e8154193e5e29c7170dc519af61f44156e7e"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-06-25 19:52:03.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-07-01 16:45:35.000000000","tz":0},"subject":"Implement flat enforcement model","message":"Implement flat enforcement model\n\nThis commit attempts to implement the flat enforcement model, where\nproject hierarchy is not taken into account during usage and limit\ncalculation and enforcement.\n\nThis commit also adds some helper methods to _BaseEnforcer to help\ncollect important service and region information.\n\nChange-Id: I3913267e472caba42ee8f724ee30d46bcdd88cf5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/f5f66d120acb65f0a66c741534a285b1e75745b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/f5f66d120acb65f0a66c741534a285b1e75745b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
