)]}'
{"id":"openstack%2Foslo.limit~669461","triplet_id":"openstack%2Foslo.limit~master~I78d3a9c14f54dd38d834f99a33c41494fa22255b","project":"openstack/oslo.limit","branch":"master","topic":"python3-train","hashtags":[],"change_id":"I78d3a9c14f54dd38d834f99a33c41494fa22255b","subject":"Add Python 3 Train unit tests","status":"MERGED","created":"2019-07-05 18:45:16.000000000","updated":"2019-07-16 22:38:12.000000000","submitted":"2019-07-16 22:38:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669461-1563316692467-0ae53db0","meta_rev_id":"e4874301b94c83d878ec136095781e34e456eb79","_number":669461,"virtual_id_number":669461,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":2,"date":"2019-07-16 22:38:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"all":[{"value":2,"date":"2019-07-16 20:30:50.000000000","_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"all":[{"value":1,"date":"2019-07-16 20:49:09.000000000","_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-16 20:49:09.000000000","updated_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"},{"updated":"2019-07-16 22:38:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8bf6b1a2e0287cbeb0655cb6cb4a80ca45b4a4a0","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-05 18:45:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38d8b5a45036ea94f5c3ee64306de2c5e69ef467","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 20:39:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/61/669461/1/check/tempest-full/ca4494a/ : SUCCESS in 1h 44m 58s\n- tempest-full-py3 http://logs.openstack.org/61/669461/1/check/tempest-full-py3/c4db1c7/ : SUCCESS in 1h 47m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/669461/1/check/openstack-tox-lower-constraints/ec482d1/ : SUCCESS in 2m 43s\n- openstack-tox-pep8 http://logs.openstack.org/61/669461/1/check/openstack-tox-pep8/ef4df34/ : SUCCESS in 3m 07s\n- openstack-tox-py27 http://logs.openstack.org/61/669461/1/check/openstack-tox-py27/28aea88/ : SUCCESS in 4m 11s\n- openstack-tox-py36 http://logs.openstack.org/61/669461/1/check/openstack-tox-py36/32da772/ : SUCCESS in 4m 33s\n- openstack-tox-py37 http://logs.openstack.org/61/669461/1/check/openstack-tox-py37/362dd8f/ : SUCCESS in 4m 44s\n- openstack-tox-docs http://logs.openstack.org/61/669461/1/check/openstack-tox-docs/317f89e/html/ : SUCCESS in 3m 04s\n- build-openstack-releasenotes http://logs.openstack.org/61/669461/1/check/build-openstack-releasenotes/b98278c/html/ : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":1},{"id":"cd6413f722802bcf01cde3bb7e612c069f09adcd","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-07-16 20:30:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4732ac026a76b8dfcbba5a1bc1469f13eef3fb2e","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-07-16 20:49:09.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"325546282ae94ee776aaec098138e75d82cad0e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 20:49:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"14c6c11569832937b3824a72ba6a0479c1fdda92","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 22:38:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full http://logs.openstack.org/61/669461/1/gate/tempest-full/1f35a53/ : SUCCESS in 1h 37m 32s\n- tempest-full-py3 http://logs.openstack.org/61/669461/1/gate/tempest-full-py3/a94d8c8/ : SUCCESS in 1h 16m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/669461/1/gate/openstack-tox-lower-constraints/7a38074/ : SUCCESS in 3m 35s\n- openstack-tox-pep8 http://logs.openstack.org/61/669461/1/gate/openstack-tox-pep8/14cdc63/ : SUCCESS in 2m 48s\n- openstack-tox-py27 http://logs.openstack.org/61/669461/1/gate/openstack-tox-py27/881d762/ : SUCCESS in 3m 38s\n- openstack-tox-py36 http://logs.openstack.org/61/669461/1/gate/openstack-tox-py36/cf91d14/ : SUCCESS in 2m 57s\n- openstack-tox-py37 http://logs.openstack.org/61/669461/1/gate/openstack-tox-py37/e77c3aa/ : SUCCESS in 3m 14s\n- openstack-tox-docs http://logs.openstack.org/61/669461/1/gate/openstack-tox-docs/9c4a9d7/html/ : SUCCESS in 2m 39s\n- build-openstack-releasenotes http://logs.openstack.org/61/669461/1/gate/build-openstack-releasenotes/4870f24/html/ : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":1},{"id":"063c9dc14c067080f6d5e93cf6f5e92acd9f4747","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 22:38:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a052897f2da740bfd346368ef7cac15aa530fc7e","revisions":{"a052897f2da740bfd346368ef7cac15aa530fc7e":{"kind":"REWORK","_number":1,"created":"2019-07-05 18:45:16.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/61/669461/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/61/669461/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/61/669461/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/61/669461/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/61/669461/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/61/669461/1"}}},"commit":{"parents":[{"commit":"90acc274f7f944e4eab4de2a114253ea24c5e6af","subject":"Rename filter_resource resource_filters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/90acc274f7f944e4eab4de2a114253ea24c5e6af"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 15:56:49.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 18:45:16.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I78d3a9c14f54dd38d834f99a33c41494fa22255b\nStory: #2005924\nTask: #34234\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/a052897f2da740bfd346368ef7cac15aa530fc7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/a052897f2da740bfd346368ef7cac15aa530fc7e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
