)]}'
{"id":"openstack%2Foslo.limit~670079","triplet_id":"openstack%2Foslo.limit~master~I50c41805b8e196fdaf20c95e9aa687b98c369a57","project":"openstack/oslo.limit","branch":"master","topic":"rewrite","hashtags":[],"change_id":"I50c41805b8e196fdaf20c95e9aa687b98c369a57","subject":"Add devstack job to .zuul.conf","status":"MERGED","created":"2019-07-10 14:44:48.000000000","updated":"2019-12-02 23:36:34.000000000","submitted":"2019-12-02 23:34:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670079-1575329675846-82587bb4","meta_rev_id":"3be63f0884c8651157c1d0199609c44bdf08b1c2","_number":670079,"virtual_id_number":670079,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-02 23:34:35.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-12-02 18:15:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-02 18:15:51.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2019-12-02 18:15:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-17 08:13:30.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-07-17 15:09:46.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2019-12-02 18:15:55.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-12-02 23:34:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8cea2c56896c253b3817fe38c11e2bdfd49b2fc5","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-07-10 14:44:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a7b8628eace952fe67590f0a853f3267a80ff18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 14:45:03.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/oslo.limit on branch master.  The error was:\n\n  Configuration item has more than one key.  Each zuul.yaml\n  configuration file must be a list of dictionaries with a\n  single key, for example:\n  \n  - job:\n      name: foo\n  \n  - project:\n      name: bar\n  \n  Ensure that every item in the list is a dictionary with only\n  one key (in this example, \u0027job\u0027 and \u0027project\u0027).  This error\n  may be caused by insufficient indentation of the keys under\n  the configuration item (\u0027name\u0027 in this example).","accounts_in_message":[],"_revision_number":1},{"id":"c567b328112e16594f2104544fb0969e424bf883","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-07-10 17:24:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ffd6f94e2ccc4b93c13f68a51fdd5fe8d0b15058","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 19:16:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/79/670079/2/check/tempest-full/f7370ab/ : SUCCESS in 1h 45m 32s\n- tempest-full-py3 http://logs.openstack.org/79/670079/2/check/tempest-full-py3/e09876f/ : SUCCESS in 1h 21m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/79/670079/2/check/openstack-tox-lower-constraints/2a910fe/ : SUCCESS in 2m 43s\n- openstack-tox-pep8 http://logs.openstack.org/79/670079/2/check/openstack-tox-pep8/7f9c227/ : SUCCESS in 2m 40s\n- openstack-tox-py27 http://logs.openstack.org/79/670079/2/check/openstack-tox-py27/d914980/ : SUCCESS in 3m 44s\n- openstack-tox-py36 http://logs.openstack.org/79/670079/2/check/openstack-tox-py36/9be9587/ : SUCCESS in 3m 09s\n- openstack-tox-py37 http://logs.openstack.org/79/670079/2/check/openstack-tox-py37/05263d9/ : SUCCESS in 3m 20s\n- openstack-tox-docs http://logs.openstack.org/79/670079/2/check/openstack-tox-docs/7b539bc/html/ : SUCCESS in 2m 54s\n- oslo-limit-functional-devstack http://logs.openstack.org/79/670079/2/check/oslo-limit-functional-devstack/8ddc204/ : SUCCESS in 38m 10s","accounts_in_message":[],"_revision_number":2},{"id":"b28aef8b45983e79c63adcde464edea7a86ee2ef","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-07-17 08:13:30.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ada78bc077dd142ff5583973c92c3c172481ecf7","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2019-07-17 15:09:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6e6ae7c115ab19a9161863e835630500aed550d0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-02 18:15:51.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"0d6001bb39569c83b98ee2bf8776e87b2469cb9a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-02 18:15:55.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a5faade13523a1248f1d71efb968617a8c5c7459","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 19:52:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/6d06e77897bd41bba509e3a2a1420454 : SUCCESS in 1h 29m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6d28fb47a96940cf9d45e229e88118aa : SUCCESS in 1h 08m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e7ed7fb194034fe8a231af8b88bf9f2d : SUCCESS in 2m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7e0fe85193b478f97555e019d82b8e2 : SUCCESS in 2m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/46fce27deb6f4420957c1ef8b05f2204 : SUCCESS in 2m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e84aec9623374030a39dd9182814406f : SUCCESS in 2m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ec80e3af4e674a7895f36b9ae9193ab4 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2e5aa248d8b4902b863ef0ee087f7ab : SUCCESS in 3m 21s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/49ff7c84c79a446f943a2d0b232c14d1 : SUCCESS in 29m 11s","accounts_in_message":[],"_revision_number":3},{"id":"6953291b387046840d07063fb3377886000ca693","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 19:52:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"77589330920d02c2f139bd66eec000ce0a1389ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 23:34:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"b78c0e9a0319df3151b5dd19a10f6a56968c55c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 23:34:35.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/c351d11eae33452caf0c7c133c0260dc : SUCCESS in 1h 34m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e9bc4df758114d1c9ceb40fd952e1d20 : SUCCESS in 1h 21m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3a52122bcb7402fb4da9dcf6fde90ed : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bce2234b77d74ff683fe4ee82fcb3367 : SUCCESS in 3m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/61476fd8b89544269fa89976f61ace77 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a6dcf82f6854a74a1bbd6d1618794f4 : SUCCESS in 3m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/115539a967b3410b9bd229d1e230b2b1 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef2a7c67308e4a4bb07e090ce640b7bf : SUCCESS in 4m 03s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/043000d811db4a7f8d02e36d4a876749 : SUCCESS in 29m 29s","accounts_in_message":[],"_revision_number":3},{"id":"18dcc001a238f9012fc387742230635af4699027","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 23:36:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5811fef82d4843df97c3bd08c07060c6 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2632c8e1a9d8ad416a364f867759e9b221f72706","revisions":{"6f20185a5931b8affe7913dbdaf12f821eb52d6f":{"kind":"REWORK","_number":1,"created":"2019-07-10 14:44:48.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/79/670079/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/79/670079/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/79/670079/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/79/670079/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/79/670079/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/79/670079/1"}}},"commit":{"parents":[{"commit":"52e2a22cd86bcd338b25d9bc9c24e8dd7d35f013","subject":"Add ksa connection logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/52e2a22cd86bcd338b25d9bc9c24e8dd7d35f013"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-07-10 14:42:43.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-07-10 14:42:43.000000000","tz":0},"subject":"Add devstack job to .zuul.conf","message":"Add devstack job to .zuul.conf\n\nWe know we want to have some level of functional testing to verify\nenforcement models. This commit uses one approach to add a base\ndevstack job to oslo.limit\u0027s .zuul.conf so that we have a minimal-ish\ninstallation to configure with limits and write actual integration and\nfunctional tests against.\n\nChange-Id: I50c41805b8e196fdaf20c95e9aa687b98c369a57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/6f20185a5931b8affe7913dbdaf12f821eb52d6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/6f20185a5931b8affe7913dbdaf12f821eb52d6f"}]},"branch":"refs/heads/master"},"6bb5d65d2dcf1c11361d940daf6108ad3f249f8e":{"kind":"REWORK","_number":2,"created":"2019-07-10 17:24:42.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/79/670079/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/79/670079/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/79/670079/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/79/670079/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/79/670079/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/79/670079/2"}}},"commit":{"parents":[{"commit":"52e2a22cd86bcd338b25d9bc9c24e8dd7d35f013","subject":"Add ksa connection logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/52e2a22cd86bcd338b25d9bc9c24e8dd7d35f013"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-07-10 14:42:43.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-07-10 17:24:33.000000000","tz":0},"subject":"Add devstack job to .zuul.conf","message":"Add devstack job to .zuul.conf\n\nWe know we want to have some level of functional testing to verify\nenforcement models. This commit uses one approach to add a base\ndevstack job to oslo.limit\u0027s .zuul.conf so that we have a minimal-ish\ninstallation to configure with limits and write actual integration and\nfunctional tests against.\n\nChange-Id: I50c41805b8e196fdaf20c95e9aa687b98c369a57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/6bb5d65d2dcf1c11361d940daf6108ad3f249f8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/6bb5d65d2dcf1c11361d940daf6108ad3f249f8e"}]},"branch":"refs/heads/master"},"2632c8e1a9d8ad416a364f867759e9b221f72706":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-12-02 18:15:51.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/79/670079/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/79/670079/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/79/670079/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/79/670079/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/79/670079/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/79/670079/3"}}},"commit":{"parents":[{"commit":"c02c025a66f1ad64a9626d9cefb962c51d5eedb7","subject":"Add ksa connection logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/c02c025a66f1ad64a9626d9cefb962c51d5eedb7"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-07-10 14:42:43.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-12-02 18:15:51.000000000","tz":0},"subject":"Add devstack job to .zuul.conf","message":"Add devstack job to .zuul.conf\n\nWe know we want to have some level of functional testing to verify\nenforcement models. This commit uses one approach to add a base\ndevstack job to oslo.limit\u0027s .zuul.conf so that we have a minimal-ish\ninstallation to configure with limits and write actual integration and\nfunctional tests against.\n\nChange-Id: I50c41805b8e196fdaf20c95e9aa687b98c369a57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/2632c8e1a9d8ad416a364f867759e9b221f72706"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/2632c8e1a9d8ad416a364f867759e9b221f72706"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
