)]}'
{"id":"openstack%2Foslo.limit~695527","triplet_id":"openstack%2Foslo.limit~master~I4d69d0982b33222977eb8c2b1a586f02a5e41cae","project":"openstack/oslo.limit","branch":"master","attention_set":{"6928":{"account":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"last_update":"2023-05-02 13:21:49.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"782":{"account":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"last_update":"2024-10-29 17:47:24.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-29 17:47:24.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I4d69d0982b33222977eb8c2b1a586f02a5e41cae","subject":"Two level limit enforcer","status":"NEW","created":"2019-11-21 16:54:08.000000000","updated":"2025-09-15 11:04:56.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":18,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"3afa61fbe79e0c3a0f2ac1e131b785592ca6a484","_number":695527,"virtual_id_number":695527,"owner":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":136,"name":"Tim Bell","email":"tim.bell@cern.ch","username":"tim-bell"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-07-12 19:16:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":-1,"date":"2024-10-26 16:48:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":136,"name":"Tim Bell","email":"tim.bell@cern.ch","username":"tim-bell"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2025-09-15 11:04:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":136,"name":"Tim Bell","email":"tim.bell@cern.ch","username":"tim-bell"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":136,"name":"Tim Bell","email":"tim.bell@cern.ch","username":"tim-bell"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":136,"name":"Tim Bell","email":"tim.bell@cern.ch","username":"tim-bell"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-21 19:37:45.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2019-12-04 11:08:09.000000000","updated_by":{"_account_id":136,"name":"Tim Bell","email":"tim.bell@cern.ch","username":"tim-bell"},"reviewer":{"_account_id":136,"name":"Tim Bell","email":"tim.bell@cern.ch","username":"tim-bell"},"state":"REVIEWER"},{"updated":"2019-12-04 11:25:49.000000000","updated_by":{"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},"reviewer":{"_account_id":2033,"name":"Belmiro Moreira","email":"moreira.belmiro.email.lists@gmail.com","username":"moreira-belmiro-email-lists"},"state":"REVIEWER"},{"updated":"2020-02-15 01:12:50.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2020-06-01 22:52:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-15 16:02:44.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2020-10-05 15:12:53.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2024-10-09 09:49:18.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2024-10-26 16:48:33.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"b4ad51518c54a62d6ae4d279bd6c35d2843626b0","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-11-21 16:54:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"551c50177e45bf4188fb5b1bdeed3e2badd1cad1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 18:47:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/3fd38f6c32b340c792a09ae15363c289 : SUCCESS in 1h 27m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d68b1563d3ee4512bafebd432ad2c0a9 : SUCCESS in 1h 43m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2255e84884c345feb019289ad5c4ca8c : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b606b42b9f142d7bc7cfca295e182c8 : FAILURE in 3m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6a9b15b6ec064fcf93efe09a30a88546 : FAILURE in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba1440f5d4fa4cdd9fb122429e8d9992 : SUCCESS in 2m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cbed4ecc389c437ca21586b218d13c49 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22d223f2bf764ae7b245b2374824db96 : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"c30a4b2d9eb36b7f70930538f737754560a017f6","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-11-22 13:57:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d9af2c4c11f0852cf9d73cc84ff8f658e2807c48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 15:38:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/5afdf0db282f45c69e5334b53592aa05 : SUCCESS in 1h 27m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/86e0c904b5034fcf8c05289064b3b277 : SUCCESS in 1h 09m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3587f3987e54c74a1e84499089f15d2 : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc70c8a368a24300b2f964cd4d79f1a1 : SUCCESS in 2m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/801e87c10e884f1f9609ab9e045e2dbf : SUCCESS in 2m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7f50441e078462e9e84a0c25838f790 : SUCCESS in 2m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/21ece9c7815343389899c09ee6ac839b : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5ec342236d14a0abf9b1da911977dab : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":2},{"id":"f545e2b6ea9f0f2785fa9846d70aee61e3fdae34","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-11-22 16:59:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d56341a41489033633e4c9ac9e93b321534f2c2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 18:52:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/2d8d3062947e483fb49560a5695659a5 : SUCCESS in 1h 48m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b989a414ed0e453685f1b1e5fdcc4c3c : SUCCESS in 1h 12m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7c69b4b76e141c49936671ca4ed0b89 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb2385830a1c45c48d8ea29ba48038bd : SUCCESS in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2a12798f96e247348f349ce295e4b983 : SUCCESS in 3m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a3e6d100aa84d2a87b5e4e3339811ad : SUCCESS in 2m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/faf67b73c2ff4c60a39f593f1e512012 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66634b39f5c647fb80cfbcab719d39b8 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":3},{"id":"61dffcdab0486cebc91c3b1c3a5a62b3b1057c7b","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-11-25 11:57:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d69e580f2841daaa68ad9849a87b19ec23580e43","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-11-25 12:23:28.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"70d418f3ca921fb8042636416d8e2b449ab5af06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 13:50:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/9e697e85cac641688e2a72063d3d6d75 : SUCCESS in 1h 16m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dfdfbe45f572471e8107de328b38da83 : SUCCESS in 1h 24m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8036c66a07564d8989f3a917094ea521 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86538b6ed3cb46db8b05059e92b12249 : SUCCESS in 3m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9fa9fc21afd14db88c382a39a61ea64f : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/722d1647bb544a009682dfb42b8d0a24 : SUCCESS in 3m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/264a620255634a56ad648ae23ef4675e : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb8bcd4facce4cb985cb8c312375c53b : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":5},{"id":"46951ee724a91072eb1064e1503aeca9ffaead03","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-11-25 18:31:07.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"b8f0d26ae77e22f828f9c588bcf1a9b3165160e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 20:15:10.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/712faf8b64b54b7789059da83e1c5d22 : SUCCESS in 1h 36m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7aa9fbb729c34f3e8be5ba6f440f7923 : SUCCESS in 1h 31m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4e2ed4a43dc4197bd69948141dff2af : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba086491747b4052af42b3a1cb935d3d : SUCCESS in 3m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cc45bc57d5154b34be8e491ab93af194 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/889f92ba2efe48c5996350754f59dd2d : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b02b15395ca540cebe9fbd128e2b107d : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c559dd95b034d6fa0c1e7ad1b033f8d : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":6},{"id":"3635fc7cecb366fa05a2b666571933f2f0a065f3","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-11-27 19:19:56.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"502f1b2dc423f37605db167b7057f541f979575d","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-12-02 10:16:01.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"82e0c3f11faed5c675f2a73c90681e4be4e737b4","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-12-18 22:39:18.000000000","message":"Patch Set 6: Workflow-1\n\n(1 comment)\n\nWIP per commit message and because it sounds like there are some outstanding questions from the inline comments.","accounts_in_message":[],"_revision_number":6},{"id":"2ebd97709cdf12f211d239ff38bd5fd5d3611d37","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-02-14 23:05:56.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8e573aa8127e4ba60ab97a60ed6dfe1bdfd8b7f7","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-02-15 01:12:50.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"831d08cae8f3c94c28fd698e74d5a7b78bcfb54b","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-06-01 21:39:54.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8d3605a61eb3029e3c044e06240421f591b6e078","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 22:52:51.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b2d827719a61421bb6f48c39dbe57eac : SUCCESS in 1h 12m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2bdb89b28704efe9c07b22321b0f35d : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adf9efbfeeaf40179e0c9a0689a19961 : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b98470d131384b1ca34bbd700cf4adb6 : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b1e1a554a2745f6ae3fb9c352c09ea6 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90540dc77fe64893805dff98748ef2c7 : SUCCESS in 3m 38s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/4e2ab6d65dd54a4abeec5cdee1d68f67 : SUCCESS in 24m 12s","accounts_in_message":[],"_revision_number":7},{"id":"ade44dc9707685fc8cf93411421229f62e1332d2","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-06-15 16:02:44.000000000","message":"Patch Set 7: Code-Review+1\n\n(2 comments)\n\nCode looks reasonable. Would like to see a bit more unit test coverage, but otherwise I\u0027m good with it, pending John\u0027s testing.","accounts_in_message":[],"_revision_number":7},{"id":"098ba874f47d1b6963865e0204b581482bd7345f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-05 15:12:53.000000000","message":"Patch Set 7:\n\nAny update?","accounts_in_message":[],"_revision_number":7},{"id":"88f0d5c99a581cb2fdd7a0e2b2d67f733ece9c4c","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-10-15 16:38:57.000000000","message":"Patch Set 7:\n\nBased on my previous comments, I think I was waiting for this to be tested with Nova. I\u0027m also not sure if anyone is still planning to push oslo.limit. I guess it\u0027s been blocked on me because of https://review.opendev.org/#/c/733881/ (which should be ready to merge now...).","accounts_in_message":[],"_revision_number":7},{"id":"e08ce1240398c480b09d64342bcb13f6d5d8f499","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-02 15:14:31.000000000","message":"Patch Set 7:\n\nHello,\n\nDo we have any update about this patch?\nDo we want to continue with these changes?","accounts_in_message":[],"_revision_number":7},{"id":"1225db418fe7535eb9098b0365bfd84b8702aeff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2023-05-02 13:21:49.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"ea8de8014e25e9df32c2b46d91b87d23862e26aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 14:38:02.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/104146204cd843629b6ef854190e5f93\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/34d8185f2f99494a87acd8b0c4681923 : SUCCESS in 1h 13m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/817ff193c68f4d859bad6be056ab2b19 : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79d83feaba35491893f33f733a3c7214 : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f021a2c2ddb243438bea74ad20edfa40 : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44eee2e4bb834e4d9d7580ed31344872 : SUCCESS in 3m 17s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/f4e65fb6a7cd461ebf95ec80793e6a4b : SUCCESS in 26m 35s","accounts_in_message":[],"_revision_number":8},{"id":"a3042c660bcc377d4b8b5a400a47379710d14d47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2023-05-02 15:10:16.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"60b0b61531abeaa302a864685c96be48ff6b4593","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2023-05-02 15:11:13.000000000","message":"Uploaded patch set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"709ad11914550c912449bc278afe27fd49885697","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2023-05-02 15:15:38.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"7beedb3b7d7626d62810ca879bb6b88d3fa26879","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2023-05-02 15:29:46.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"5d1e159cddd9835b06ad94199820e624b9faf902","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 16:54:15.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1aff163d2cf04cd28abf6929cf07b060\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ebe22894d39b4f9d8dc2e0872504fd0d : SUCCESS in 1h 18m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90825e1450e84013b7ac06b731b43fe3 : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28f7f6ae4646479bb32e8e54c54b129d : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0cf878b45a314e27beebaa667e01b08b : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dbf0ba85c6241ac888f785c11bc544e : SUCCESS in 2m 46s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/b30390c5982c45e88dadaaa676f5d60b : SUCCESS in 29m 40s","accounts_in_message":[],"_revision_number":11},{"id":"509dc1779e1113ad28435a91614db5bb5e89afe1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2023-05-02 17:21:18.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"ee64b475d98edfe75fd2963ffd10bd428e4a98e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 18:44:45.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c4f3aaf0c424c14a78c3bb8fcee2f9d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/77bd395b3b054c3e93fa1f049d77af1d : SUCCESS in 1h 21m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0d5175a9393482994e354a64b1f000e : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f70f8ceca75a4fa9901c1577c79ae740 : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/86d39aa8ca834dccabc554cdb52eb536 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f51da6d4e954f5980ceea2b2bf0b4e7 : SUCCESS in 3m 08s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/0871e87ee7d34d4198f3d8e9a14a87a1 : SUCCESS in 31m 47s","accounts_in_message":[],"_revision_number":12},{"id":"9df5321c7f731784c2a369f3c8e868db9cfe999c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2023-10-06 16:55:19.000000000","message":"Patch Set 13: Patch Set 12 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"cff4f10decd443a71f4a1b7d352d6439c01bf1a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 18:38:52.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dba0f338594464bb540b94ee1ae1cf5\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1a87c5b4b4214b6cbc76c743596a1b15 : SUCCESS in 1h 36m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f54881c43f34909945ffd4649ab3f83 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e1a3cb82d2748458bdae3665c7b2c66 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9c684ae56704037adc681d14e597811 : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ddaa6e05434c495b88a6b10475d33ec0 : SUCCESS in 3m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4c0d13a2dde145ce90318b6fc3a902fd : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/479b427738f3423f9cdbfaa17797c037 : SUCCESS in 3m 31s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/5dc8374c078447fa9172014ab4c27c6a : SUCCESS in 20m 10s","accounts_in_message":[],"_revision_number":13},{"id":"286864abb0475330693544093143555c43177d62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2024-04-10 09:45:46.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"ed613825147ca5e2b64ee71f02da8e3c478c6d08","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2024-04-10 09:47:21.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"8b839ad8ef6418a7f6c29008f9e4cfb3e66d1392","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 11:14:16.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5051f90552d545d489afaca0034d0bb0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16fcd6f41c244ae7a129b9cea7961f87 : SUCCESS in 4m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d7f8ee20d3014353bacb3f25cbd3f08a : SUCCESS in 1h 26m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/912e0831d4f642dfa0db0e03cca37657 : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bca93e2fcb554e049f6760bddf774afc : SUCCESS in 3m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9a218681348443728c8a0d6fb6935026 : SUCCESS in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e2664846d1d345ea8cbccf685298385b : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b9b68f4221945f4a1d3246ca5618e05 : SUCCESS in 2m 59s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/b926b474949d46bdae90ff0bdc5fecca : SUCCESS in 23m 13s","accounts_in_message":[],"_revision_number":14},{"id":"c0f5c886eaa72bf7515843233a3aef498100444d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2024-06-07 11:17:20.000000000","message":"Patch Set 15: Patch Set 14 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"9675e5569a280e80dc2513bdc553bf9ba2e12ae0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-07 13:27:37.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc4e152dc6ab4aef9b01a1c2823ccf8a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57e7579a7beb4af79a97ab55a45279c1 : SUCCESS in 2m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1c44f7462f2644cfa2fd6e8411e95c2b : TIMED_OUT in 2h 03m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ccce3adc4b3404aa9b1cb5b62f9278c : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6335ca913cd542529896106867095e3b : SUCCESS in 2m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/41987d5ccc564f2aa390fb6a01649074 : SUCCESS in 2m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/868d4f55908640eca968fb30e9546eed : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15b4f2ec046648fcba2f1a612ad7ddcd : SUCCESS in 2m 48s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/c8759210694c4630b8386738b9fd2a91 : SUCCESS in 30m 27s","accounts_in_message":[],"_revision_number":15},{"id":"783e3cadce065466df0314e33ed9421d6f8baf99","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-07-12 17:42:15.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"b217c5479d7bade75d464099300adbc5d6d35730","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 19:16:14.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6a840534c7141d58d141c1fbb0b9875\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8fb35748b7824b83a6588477271c57be : SUCCESS in 2m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e6580635395042baa212b88e3b20c3a6 : SUCCESS in 1h 33m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8c5971aa5e4448fa89465868247c7b3 : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5da7c7ccc6ef4d25bdde4fd9863884ca : SUCCESS in 2m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c13be3a2e67e4e8dac758a89566e63db : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b89272b33d684b868b583bd9495b24c9 : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd9482e4f3bc49139566a3c15f32c450 : SUCCESS in 2m 52s\n- oslo-limit-functional-devstack https://zuul.opendev.org/t/openstack/build/1c98b7f4d15d4f1ca72a4468c8030276 : SUCCESS in 15m 50s","accounts_in_message":[],"_revision_number":15},{"id":"deecd0067cba5a28d81ef6e313a4eb53dd81b38c","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-10-09 09:49:18.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"b41c1b4a321a2e83dd3d4c09aa0f83b886506e9a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-26 16:48:33.000000000","message":"Patch Set 15: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"0708c81957a426877cc643d26109420bd3cca80f","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2024-10-27 21:47:55.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"bcdfcd70613389967e763d01d272ae26d9647c3d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-29 17:47:24.000000000","message":"Patch Set 15:\n\n(4 comments)","accounts_in_message":[],"_revision_number":15},{"id":"3afa61fbe79e0c3a0f2ac1e131b785592ca6a484","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2025-09-15 11:04:56.000000000","message":"Patch Set 15: -Code-Review","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"20224481172d8a20b1ac1fa8b0f8d46f391cbdd4","revisions":{"95e22691278065edef7aef1a70d17277233b23b5":{"kind":"REWORK","_number":1,"created":"2019-11-21 16:54:08.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/27/695527/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/27/695527/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/1"}}},"commit":{"parents":[{"commit":"c0f480b11b1ad6b1e7658abdd199191bd3522e42","subject":"Add flat enforcer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/c0f480b11b1ad6b1e7658abdd199191bd3522e42"}]}],"author":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2019-11-21 16:52:04.000000000","tz":0},"committer":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2019-11-21 16:52:04.000000000","tz":0},"subject":"WIP: Two level limit enforcer","message":"WIP: Two level limit enforcer\n\nBuilding on the flat enforce, added POC for two lovel enforcement.\n\nClearly needs more testing to shake out issues, but it shows how it\ncould work in a way that when there is no hierarchy its almost as\nefficient as the flat enforcer. We need extra class to keystone about\nthe project hierarchy, but otherwise no extra counting of resources.\n\nChange-Id: I4d69d0982b33222977eb8c2b1a586f02a5e41cae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/95e22691278065edef7aef1a70d17277233b23b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/95e22691278065edef7aef1a70d17277233b23b5"}]},"branch":"refs/heads/master"},"044724a3c0d0eaca8c4366d31890ae35b8883d6d":{"kind":"REWORK","_number":2,"created":"2019-11-22 13:57:50.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/27/695527/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/27/695527/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/2"}}},"commit":{"parents":[{"commit":"921c448389cb2e844fb34abd8db7f7a984ca3fa8","subject":"Add flat enforcer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/921c448389cb2e844fb34abd8db7f7a984ca3fa8"}]}],"author":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2019-11-21 16:52:04.000000000","tz":0},"committer":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2019-11-22 13:57:40.000000000","tz":0},"subject":"WIP: Two level limit enforcer","message":"WIP: Two level limit enforcer\n\nBuilding on the flat enforce, added POC for two lovel enforcement.\n\nClearly needs more testing to shake out issues, but it shows how it\ncould work in a way that when there is no hierarchy its almost as\nefficient as the flat enforcer. We need extra class to keystone about\nthe project hierarchy, but otherwise no extra counting of resources.\n\nChange-Id: I4d69d0982b33222977eb8c2b1a586f02a5e41cae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/044724a3c0d0eaca8c4366d31890ae35b8883d6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/044724a3c0d0eaca8c4366d31890ae35b8883d6d"}]},"branch":"refs/heads/master"},"74f650f05d1fbd514b6f8018e76c1f2d7309918a":{"kind":"REWORK","_number":3,"created":"2019-11-22 16:59:18.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/27/695527/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/27/695527/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/3"}}},"commit":{"parents":[{"commit":"dd70d1a331efcbc0eaf44134739d0ace8fa94b50","subject":"Add flat enforcer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/dd70d1a331efcbc0eaf44134739d0ace8fa94b50"}]}],"author":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2019-11-22 16:56:19.000000000","tz":0},"committer":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2019-11-22 16:59:08.000000000","tz":0},"subject":"WIP: Two level limit enforcer","message":"WIP: Two level limit enforcer\n\nBuilding on the flat enforce, added POC for two lovel enforcement.\n\nClearly needs more testing to shake out issues, but it shows how it\ncould work in a way that when there is no hierarchy its almost as\nefficient as the flat enforcer. We need extra class to keystone about\nthe project hierarchy, but otherwise no extra counting of resources.\n\nChange-Id: I4d69d0982b33222977eb8c2b1a586f02a5e41cae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/74f650f05d1fbd514b6f8018e76c1f2d7309918a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/74f650f05d1fbd514b6f8018e76c1f2d7309918a"}]},"branch":"refs/heads/master"},"9e9a972d4807e450f47b79e81671487073832967":{"kind":"REWORK","_number":4,"created":"2019-11-25 11:57:31.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/27/695527/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/27/695527/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/4"}}},"commit":{"parents":[{"commit":"94a735863694b3e78ac720b37629c16fbc5bba3a","subject":"Add flat enforcer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/94a735863694b3e78ac720b37629c16fbc5bba3a"}]}],"author":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2019-11-22 16:56:19.000000000","tz":0},"committer":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2019-11-25 11:56:54.000000000","tz":0},"subject":"WIP: Two level limit enforcer","message":"WIP: Two level limit enforcer\n\nBuilding on the flat enforce, added POC for two lovel enforcement.\n\nClearly needs more testing to shake out issues, but it shows how it\ncould work in a way that when there is no hierarchy its almost as\nefficient as the flat enforcer. We need extra class to keystone about\nthe project hierarchy, but otherwise no extra counting of resources.\n\nChange-Id: I4d69d0982b33222977eb8c2b1a586f02a5e41cae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/9e9a972d4807e450f47b79e81671487073832967"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/9e9a972d4807e450f47b79e81671487073832967"}]},"branch":"refs/heads/master"},"3fdb9abd551b11e7d4ab67e15c3f48efe3ecbb2a":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-11-25 12:23:28.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/27/695527/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/27/695527/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/5"}}},"commit":{"parents":[{"commit":"47144ab60449b2b4cff5346778fa92bf3553edb5","subject":"Add flat enforcer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/47144ab60449b2b4cff5346778fa92bf3553edb5"}]}],"author":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2019-11-22 16:56:19.000000000","tz":0},"committer":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2019-11-25 12:23:28.000000000","tz":0},"subject":"WIP: Two level limit enforcer","message":"WIP: Two level limit enforcer\n\nBuilding on the flat enforce, added POC for two lovel enforcement.\n\nClearly needs more testing to shake out issues, but it shows how it\ncould work in a way that when there is no hierarchy its almost as\nefficient as the flat enforcer. We need extra class to keystone about\nthe project hierarchy, but otherwise no extra counting of resources.\n\nChange-Id: I4d69d0982b33222977eb8c2b1a586f02a5e41cae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/3fdb9abd551b11e7d4ab67e15c3f48efe3ecbb2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/3fdb9abd551b11e7d4ab67e15c3f48efe3ecbb2a"}]},"branch":"refs/heads/master"},"0a54c217286ec26b18d97a554c0b0104142992c8":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-11-25 18:31:07.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/27/695527/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/27/695527/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/6"}}},"commit":{"parents":[{"commit":"40ef2764a5b8483103ce4d7b8d390b4e5d643d79","subject":"Add flat enforcer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/40ef2764a5b8483103ce4d7b8d390b4e5d643d79"}]}],"author":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2019-11-22 16:56:19.000000000","tz":0},"committer":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2019-11-25 18:30:58.000000000","tz":0},"subject":"WIP: Two level limit enforcer","message":"WIP: Two level limit enforcer\n\nBuilding on the flat enforce, added POC for two lovel enforcement.\n\nClearly needs more testing to shake out issues, but it shows how it\ncould work in a way that when there is no hierarchy its almost as\nefficient as the flat enforcer. We need extra class to keystone about\nthe project hierarchy, but otherwise no extra counting of resources.\n\nChange-Id: I4d69d0982b33222977eb8c2b1a586f02a5e41cae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/0a54c217286ec26b18d97a554c0b0104142992c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/0a54c217286ec26b18d97a554c0b0104142992c8"}]},"branch":"refs/heads/master"},"46c4f026e5e8cac6d5581e5909f500edca98ceff":{"kind":"REWORK","_number":7,"created":"2020-06-01 21:39:54.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/27/695527/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/27/695527/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/7"}}},"commit":{"parents":[{"commit":"1a64aa8dd58dc9a2d83c1b5e6d86a76d74b51ce2","subject":"Merge \"Remove translation sections from setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/1a64aa8dd58dc9a2d83c1b5e6d86a76d74b51ce2"}]}],"author":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2019-11-22 16:56:19.000000000","tz":0},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.com","date":"2020-06-01 21:39:14.000000000","tz":-420},"subject":"Two level limit enforcer","message":"Two level limit enforcer\n\nBuilding on the flat enforce, added POC for two lovel enforcement.\n\nClearly needs more testing to shake out issues, but it shows how it\ncould work in a way that when there is no hierarchy its almost as\nefficient as the flat enforcer. We need extra class to keystone about\nthe project hierarchy, but otherwise no extra counting of resources.\n\nCo-authored-by: Colleen Murphy \u003ccolleen.murphy@suse.com\u003e\n\nChange-Id: I4d69d0982b33222977eb8c2b1a586f02a5e41cae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/46c4f026e5e8cac6d5581e5909f500edca98ceff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/46c4f026e5e8cac6d5581e5909f500edca98ceff"}]},"branch":"refs/heads/master"},"492c9b3e096fb0d209e572ec06b645dead9d33d9":{"kind":"REWORK","_number":8,"created":"2023-05-02 13:21:49.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/27/695527/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/27/695527/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/8"}}},"commit":{"parents":[{"commit":"886e7170f1d543da37de738660efee0a01cf1bac","subject":"Moves supported python runtimes from version 3.8 to 3.10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/886e7170f1d543da37de738660efee0a01cf1bac"}]}],"author":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2019-11-22 16:56:19.000000000","tz":0},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2023-05-02 13:21:50.000000000","tz":60},"subject":"Two level limit enforcer","message":"Two level limit enforcer\n\nBuilding on the flat enforce, added POC for two lovel enforcement.\n\nClearly needs more testing to shake out issues, but it shows how it\ncould work in a way that when there is no hierarchy its almost as\nefficient as the flat enforcer. We need extra class to keystone about\nthe project hierarchy, but otherwise no extra counting of resources.\n\nCo-authored-by: Colleen Murphy \u003ccolleen.murphy@suse.com\u003e\n\nChange-Id: I4d69d0982b33222977eb8c2b1a586f02a5e41cae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/492c9b3e096fb0d209e572ec06b645dead9d33d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/492c9b3e096fb0d209e572ec06b645dead9d33d9"}]},"branch":"refs/heads/master"},"3b7c3215c80eb30c0c9571a85ca440a074099a14":{"kind":"REWORK","_number":9,"created":"2023-05-02 15:10:16.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/27/695527/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/27/695527/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/9"}}},"commit":{"parents":[{"commit":"9d9c96e5c224c25e0ff2680f1bb041a5aea601b1","subject":"Fix up tox -e cover to use coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/9d9c96e5c224c25e0ff2680f1bb041a5aea601b1"}]}],"author":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2019-11-22 16:56:19.000000000","tz":0},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2023-05-02 15:09:20.000000000","tz":60},"subject":"Two level limit enforcer","message":"Two level limit enforcer\n\nBuilding on the flat enforce, added POC for two lovel enforcement.\n\nClearly needs more testing to shake out issues, but it shows how it\ncould work in a way that when there is no hierarchy its almost as\nefficient as the flat enforcer. We need extra class to keystone about\nthe project hierarchy, but otherwise no extra counting of resources.\n\nCo-authored-by: Colleen Murphy \u003ccolleen.murphy@suse.com\u003e\n\nChange-Id: I4d69d0982b33222977eb8c2b1a586f02a5e41cae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/3b7c3215c80eb30c0c9571a85ca440a074099a14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/3b7c3215c80eb30c0c9571a85ca440a074099a14"}]},"branch":"refs/heads/master"},"544a8f863c85c3d755eb30b854ed554a40c99629":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2023-05-02 15:11:13.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/27/695527/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/27/695527/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/10"}}},"commit":{"parents":[{"commit":"9d9c96e5c224c25e0ff2680f1bb041a5aea601b1","subject":"Fix up tox -e cover to use coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/9d9c96e5c224c25e0ff2680f1bb041a5aea601b1"}]}],"author":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2019-11-22 16:56:19.000000000","tz":0},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2023-05-02 15:11:01.000000000","tz":60},"subject":"Two level limit enforcer","message":"Two level limit enforcer\n\nBuilding on the flat enforce, added POC for two level enforcement.\n\nClearly needs more testing to shake out issues, but it shows how it\ncould work in a way that when there is no hierarchy its almost as\nefficient as the flat enforcer. We need extra class to keystone about\nthe project hierarchy, but otherwise no extra counting of resources.\n\nCo-authored-by: Colleen Murphy \u003ccolleen.murphy@suse.com\u003e\n\nChange-Id: I4d69d0982b33222977eb8c2b1a586f02a5e41cae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/544a8f863c85c3d755eb30b854ed554a40c99629"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/544a8f863c85c3d755eb30b854ed554a40c99629"}]},"branch":"refs/heads/master"},"6e573d5dcd11984805af366b389eef40fa5f158e":{"kind":"REWORK","_number":11,"created":"2023-05-02 15:29:46.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/27/695527/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/27/695527/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/11"}}},"commit":{"parents":[{"commit":"9d9c96e5c224c25e0ff2680f1bb041a5aea601b1","subject":"Fix up tox -e cover to use coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/9d9c96e5c224c25e0ff2680f1bb041a5aea601b1"}]}],"author":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2019-11-22 16:56:19.000000000","tz":0},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2023-05-02 15:29:43.000000000","tz":60},"subject":"Two level limit enforcer","message":"Two level limit enforcer\n\nBuilding on the flat enforce, added POC for two level enforcement.\n\nClearly needs more testing to shake out issues, but it shows how it\ncould work in a way that when there is no hierarchy its almost as\nefficient as the flat enforcer. We need extra class to keystone about\nthe project hierarchy, but otherwise no extra counting of resources.\n\nCo-authored-by: Colleen Murphy \u003ccolleen.murphy@suse.com\u003e\n\nChange-Id: I4d69d0982b33222977eb8c2b1a586f02a5e41cae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/6e573d5dcd11984805af366b389eef40fa5f158e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/6e573d5dcd11984805af366b389eef40fa5f158e"}]},"branch":"refs/heads/master"},"ecb1c9f8b16b82315b5f37b076b62a35830a1520":{"kind":"REWORK","_number":12,"created":"2023-05-02 17:21:18.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/27/695527/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/27/695527/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/12"}}},"commit":{"parents":[{"commit":"9d9c96e5c224c25e0ff2680f1bb041a5aea601b1","subject":"Fix up tox -e cover to use coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/9d9c96e5c224c25e0ff2680f1bb041a5aea601b1"}]}],"author":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2019-11-22 16:56:19.000000000","tz":0},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2023-05-02 17:15:19.000000000","tz":60},"subject":"Two level limit enforcer","message":"Two level limit enforcer\n\nBuilding on the flat enforcer, added two level enforcement.\n\nIf project A has children B and C, we enforce that any\nlimits on project A are applied to usage across A, B and C.\n\nIn addition, when add resources to a child project we\nfirst ensure limits on the child project usage are not exceeded.\nWe then find the parent project (project A) and check the\nusage across all projects in the tree (A, B and C) are not\nexcceeded.\n\nNote, that project A might be in the default domain, so\nkeystone will say the parent of project A is the default\ndomain, but here we use project A the root of\nthe enforcement tree.\n\nThis has been tested with Nova\u0027s unified limits implemenation\n(added in Yoga).\n\nCo-authored-by: Colleen Murphy \u003ccolleen.murphy@suse.com\u003e\n\nChange-Id: I4d69d0982b33222977eb8c2b1a586f02a5e41cae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/ecb1c9f8b16b82315b5f37b076b62a35830a1520"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/ecb1c9f8b16b82315b5f37b076b62a35830a1520"}]},"branch":"refs/heads/master"},"8816b5c905a69d882463e59f286c52a491953e12":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2023-10-06 16:55:19.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/27/695527/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/27/695527/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/13"}}},"commit":{"parents":[{"commit":"027d5d263cc03409a06ad114925d95e488c8ca1c","subject":"Fix up tox -e cover to use coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/027d5d263cc03409a06ad114925d95e488c8ca1c"}]}],"author":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2019-11-22 16:56:19.000000000","tz":0},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2023-10-06 16:55:19.000000000","tz":0},"subject":"Two level limit enforcer","message":"Two level limit enforcer\n\nBuilding on the flat enforcer, added two level enforcement.\n\nIf project A has children B and C, we enforce that any\nlimits on project A are applied to usage across A, B and C.\n\nIn addition, when add resources to a child project we\nfirst ensure limits on the child project usage are not exceeded.\nWe then find the parent project (project A) and check the\nusage across all projects in the tree (A, B and C) are not\nexcceeded.\n\nNote, that project A might be in the default domain, so\nkeystone will say the parent of project A is the default\ndomain, but here we use project A the root of\nthe enforcement tree.\n\nThis has been tested with Nova\u0027s unified limits implemenation\n(added in Yoga).\n\nCo-authored-by: Colleen Murphy \u003ccolleen.murphy@suse.com\u003e\n\nChange-Id: I4d69d0982b33222977eb8c2b1a586f02a5e41cae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/8816b5c905a69d882463e59f286c52a491953e12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/8816b5c905a69d882463e59f286c52a491953e12"}]},"branch":"refs/heads/master","description":"Rebase"},"036839887db592e0b4468f63c288b856fcb97ac3":{"kind":"REWORK","_number":14,"created":"2024-04-10 09:45:46.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/27/695527/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/27/695527/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/14"}}},"commit":{"parents":[{"commit":"9d645d9713cbfe81e782b8be369fcba0c3320b78","subject":"Merge \"reno: Update master for unmaintained/xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/9d645d9713cbfe81e782b8be369fcba0c3320b78"}]}],"author":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2019-11-22 16:56:19.000000000","tz":0},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2024-04-10 09:45:39.000000000","tz":60},"subject":"Two level limit enforcer","message":"Two level limit enforcer\n\nBuilding on the flat enforcer, added two level enforcement.\n\nIf project A has children B and C, we enforce that any\nlimits on project A are applied to usage across A, B and C.\n\nIn addition, when add resources to a child project we\nfirst ensure limits on the child project usage are not exceeded.\nWe then find the parent project (project A) and check the\nusage across all projects in the tree (A, B and C) are not\nexcceeded.\n\nNote, that project A might be in the default domain, so\nkeystone will say the parent of project A is the default\ndomain, but here we use project A the root of\nthe enforcement tree.\n\nThis has been tested with Nova\u0027s unified limits implemenation\n(added in Yoga).\n\nCo-authored-by: Colleen Murphy \u003ccolleen.murphy@suse.com\u003e\n\nChange-Id: I4d69d0982b33222977eb8c2b1a586f02a5e41cae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/036839887db592e0b4468f63c288b856fcb97ac3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/036839887db592e0b4468f63c288b856fcb97ac3"}]},"branch":"refs/heads/master"},"20224481172d8a20b1ac1fa8b0f8d46f391cbdd4":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2024-06-07 11:17:20.000000000","uploader":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"ref":"refs/changes/27/695527/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.limit","ref":"refs/changes/27/695527/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.limit refs/changes/27/695527/15"}}},"commit":{"parents":[{"commit":"25f58f919cf91ef72ca5272d42541aa0e503a48b","subject":"reno: Update master for unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/25f58f919cf91ef72ca5272d42541aa0e503a48b"}]}],"author":{"name":"John Garbutt","email":"john@johngarbutt.com","date":"2019-11-22 16:56:19.000000000","tz":0},"committer":{"name":"John Garbutt","email":"john.garbutt@stackhpc.com","date":"2024-06-07 11:17:20.000000000","tz":0},"subject":"Two level limit enforcer","message":"Two level limit enforcer\n\nBuilding on the flat enforcer, added two level enforcement.\n\nIf project A has children B and C, we enforce that any\nlimits on project A are applied to usage across A, B and C.\n\nIn addition, when add resources to a child project we\nfirst ensure limits on the child project usage are not exceeded.\nWe then find the parent project (project A) and check the\nusage across all projects in the tree (A, B and C) are not\nexcceeded.\n\nNote, that project A might be in the default domain, so\nkeystone will say the parent of project A is the default\ndomain, but here we use project A the root of\nthe enforcement tree.\n\nThis has been tested with Nova\u0027s unified limits implemenation\n(added in Yoga).\n\nCo-authored-by: Colleen Murphy \u003ccolleen.murphy@suse.com\u003e\n\nChange-Id: I4d69d0982b33222977eb8c2b1a586f02a5e41cae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/20224481172d8a20b1ac1fa8b0f8d46f391cbdd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.limit/commit/20224481172d8a20b1ac1fa8b0f8d46f391cbdd4"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
